)]}'
{"id":"openstack%2Frequirements~998098","triplet_id":"openstack%2Frequirements~stable%2F2025.1~If208ebff841bbb3c1feefabcce2b08aae13bf0d6","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","hashtags":[],"change_id":"If208ebff841bbb3c1feefabcce2b08aae13bf0d6","subject":"update constraint for os-ken to new release 3.0.2","status":"MERGED","created":"2026-07-21 09:19:19.000000000","updated":"2026-07-21 14:44:11.000000000","submitted":"2026-07-21 14:44:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998098-new-release","meta_rev_id":"3902ca5f09251dd4eced60bab9a58c9fabb01045","_number":998098,"virtual_id_number":998098,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-21 14:44:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-07-21 12:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-07-21 12:50:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-21 11:06:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-21 12:50:58.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"562cd74365106a09b0b5e5065f8ad027fca08fb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-21 09:19:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bb6cfb592526feb327e6356c3f8ef15918ef977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 11:06:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c928d17597644e65bed12ad1bd067b6e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffceae378e494f2ba497152817b9fbe0 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dd2e796404c41c4b4662b73281dcf88 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/afb13806fe014a84878c35c22f53ab05 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf75ee9d7a144165919185b268ab2c6b : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40bbb15a7ac5471baa5b11821d81c077 : SUCCESS in 3m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/01e26c323c2e45bd9e5a57576fb1f6fd : SUCCESS in 17m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3e23c1c9e71746d18c166bff59c7e123 : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/782c9ae7378a410da0615a9390dff74f : SUCCESS in 18m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e969342630fb42fb9fa5d14f0fc9fab2 : SUCCESS in 20m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f54fa42b43fe4ac28fde396438184b38 : SUCCESS in 5m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b12aa7a41b50426fbde2dc8cc1a4eaf1 : SUCCESS in 5m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6c7c39979e1e489ba3f4a72f7b847bcb : SUCCESS in 4m 12s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1da8e30409cb4e8ab86d4de04153646d : SUCCESS in 11m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6a744ee79f48478cbf1c9f13256961f0 : SUCCESS in 4m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6d694551ddcf4b50a83ec69f98193fc7 : SUCCESS in 5m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d7c5c2ebeccb4415bf5038ea567dfbe2 : SUCCESS in 8m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b6104e5d7ad941f8b23c2a3269c1a372 : SUCCESS in 14m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ed52347b88c4be09726499ea31e4701 : SUCCESS in 7m 20s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a035fde0047c41199f8eb6478ce60655 : SUCCESS in 10m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/effd67bcdeda4d838cac979474fdd69d : SUCCESS in 13m 19s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b54987fb0fdb45db81fc524192e6aff7 : SUCCESS in 2m 35s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/69fd7de6134540f4aeae8246caeb3304 : SUCCESS in 9m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2cc3ed3e915043988d12d08f10243907 : SUCCESS in 4m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dd1fb35eff8e45208683ed96d9f3247b : FAILURE in 4m 44s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/101f7c0abf114c0d9de83212825d1c14 : SUCCESS in 11m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1784a87b519f45a0bc1023ac0ad10f9c : SUCCESS in 20m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dbb5a2f6a88f4fd7b5c69bfbd7ffd9e3 : SUCCESS in 7m 42s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/00f1a2633ca246538ee721ede20aacac : SUCCESS in 14m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/374c951476e1445b9812e0c451097ea8 : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/67b85ec1efa0435a88d4c50ad4a63b58 : SUCCESS in 5m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38ca49b093b64458a2174a4d6b70f326 : SUCCESS in 27m 59s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/14c5748f5dc945f0b67c0e575074b970 : SUCCESS in 6m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ffdfeb25d0384c688a307fdd08699ff4 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff347dfcc2b348abb4ba4e52aae28273 : SUCCESS in 5m 40s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1c811fc285dc4f8c927079ab7ad27480 : SUCCESS in 15m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75347bb4b4114ff9848e6accbb36226b : SUCCESS in 1h 03m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77948ad496c0468faa6744ab280ea62d : SUCCESS in 1h 40m 53s","accounts_in_message":[],"_revision_number":1},{"id":"03ca9ff7ec90b9b6abb75c1eba54dfb12c79e19a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-21 12:50:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a4c12428384717be8508d91bcd28fe5f0fe9cf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 12:51:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1429127eeaae818b2ce9bd7a918e93ea7be63d2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:44:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/533c650c71644996bdfa5c29006ec621\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d0158e3ee9449e915515f49e82010c : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f9635f971c745e08c627230f9a8db70 : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38447bc9a295410e9a68bf1753ccb246 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b4904a8254245d0815a900db706ab18 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf3cd7c39fa0403fad13210b9fe71d65 : SUCCESS in 4m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/88d96da416724f0aa70b82114cc66dbe : SUCCESS in 9m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fffabcee7cfa4164aeacedb4f0a34ccd : SUCCESS in 10m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b8860a72589d4c0f9982536fa0f9fec1 : SUCCESS in 18m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7b09980256aa4daf9ce4dd267812f343 : SUCCESS in 20m 09s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c00209e88ead406d8e6a1ea7dd647f5d : SUCCESS in 9m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a0caeb2a52874593b382fbdb7d896529 : SUCCESS in 5m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ea9da025e8d74704a5ce03efaf92010f : SUCCESS in 7m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c81def7e642d48c4bb7e732906eed0ff : SUCCESS in 8m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a4af39e81ec7487cbcab807827beab45 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/caafbd09da2b4472ad495573cc55e3e5 : SUCCESS in 5m 18s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2844a168d8de4cf2af3b0f0428dbf8a6 : SUCCESS in 5m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/595f3d564d3c4a5b9b0acb5e1c6c4ca2 : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/057457bb7c434f4995cd3dbe512376c9 : SUCCESS in 9m 48s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/12c96666cdd847c7be2ca9bb868ded14 : SUCCESS in 18m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f1792634e2d745d39104e4bfacf08eac : SUCCESS in 8m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6f17d5f96c8b403aa4cd38f3b66ebc2d : SUCCESS in 16m 15s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f6e6c7b29ab74fa1b0b6a7982198875a : SUCCESS in 7m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/70893f7d53e142bc843a0069c8ae0891 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a30ffcc0bbb54cc4b63f6503ed6b3645 : SUCCESS in 29m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5eeef745031c4a839d7e7c9b9fc12dbb : SUCCESS in 4m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7f2dabaf2c19459e8b2a02665cbfa53a : SUCCESS in 8m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/519f0d77013e4c74a9b2ff786bb5c264 : SUCCESS in 19m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfecf6e27ad649c59964636c7abc9805 : SUCCESS in 59m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ad2acf136f5433cb5cecff1bd801624 : SUCCESS in 1h 47m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3902ca5f09251dd4eced60bab9a58c9fabb01045","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:44:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b30a1b6d176ddf775923debc0085fb67d70fba9f","revisions":{"b30a1b6d176ddf775923debc0085fb67d70fba9f":{"kind":"REWORK","_number":1,"created":"2026-07-21 09:19:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/998098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/998098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/998098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/998098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/998098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/998098/1"}}},"commit":{"parents":[{"commit":"316aa24519e7a317073dab1bcbdc2b2f5b4f04ce","subject":"Merge \"update constraint for oslo.config to new release 9.7.2\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/316aa24519e7a317073dab1bcbdc2b2f5b4f04ce"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-21 09:09:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-21 09:09:16.000000000","tz":0},"subject":"update constraint for os-ken to new release 3.0.2","message":"update constraint for os-ken to new release 3.0.2\n\nmeta: version: 3.0.2\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I64b29ee4225c582702e35e3eab22b3b9c0a020dc\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If208ebff841bbb3c1feefabcce2b08aae13bf0d6\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b30a1b6d176ddf775923debc0085fb67d70fba9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b30a1b6d176ddf775923debc0085fb67d70fba9f"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
