)]}'
{"id":"openstack%2Frequirements~999599","triplet_id":"openstack%2Frequirements~stable%2F2025.1~Ia9940619f6e6654d8dccac6f3be144474c8a2504","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","hashtags":[],"change_id":"Ia9940619f6e6654d8dccac6f3be144474c8a2504","subject":"update constraint for python-manilaclient to new release 5.4.1","status":"MERGED","created":"2026-08-03 12:53:08.000000000","updated":"2026-08-03 16:47:08.000000000","submitted":"2026-08-03 16:47:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999599-new-release","meta_rev_id":"219939ee9d18c76aec14a5e5182dde48269d7e02","_number":999599,"virtual_id_number":999599,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-03 16:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-08-03 15:14:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-08-03 15:14:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-03 14:18:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-03 15:14:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"b11fc9391700c81549b67a59ab8f115e407d293a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-03 12:53:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a467e6559592c24d52bc4ee272525163d62535bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 14:18:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c5f30e572a24b88bc3156671f4a34df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3469876da1429581d91b362e8d6785 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5cf3255353a4796ae9c07c5265459bf : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f6a8fdd20e747418b4f2769f819bc62 : SUCCESS in 3m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4bdf87a65124c5181a565e3bf81c83e : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c99b2d3c3d4d4af0ab798dd600b51266 : SUCCESS in 2m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/20a57babf1ae4a8bae4d6c2177bb8e7c : SUCCESS in 16m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0a678524dcac42ccb09f203f5db78a5d : SUCCESS in 10m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e532f9efd6b54fe3b0248dc41d22f81b : SUCCESS in 18m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/63d0cd8f234740fa813670965c1650ee : SUCCESS in 22m 09s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7fb1cb30d9454d00bcefc345d2becdf9 : SUCCESS in 3m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/09c4186cc78445f9b55eb43c0108ceb3 : SUCCESS in 4m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/19d453d834574be9972dedac06dfe2e1 : SUCCESS in 3m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0419349220e248dfaa19923152fd127b : SUCCESS in 12m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3c10508d9bd947f1965afea6ea3d8bad : SUCCESS in 4m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9560a7bac9c24f19951c9057eff8c888 : SUCCESS in 6m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a24b108d3cfe442b82da3f527ab6ab97 : SUCCESS in 5m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dd5c0cb1e47b43fdb35145edcddb9c7c : SUCCESS in 14m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8ffc935778264e9fa2e269f9b1744439 : SUCCESS in 5m 11s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2c47f98a296240da86d2e7e7bea7eed9 : SUCCESS in 8m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e81d8b3788b84b4cb8fc09252c7cba51 : SUCCESS in 7m 18s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d61382da02124fc1b7e94cc7fe299416 : SUCCESS in 2m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c1ae04e3c0284bc4bb1f05b4ef2fa090 : SUCCESS in 7m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6241b08211e14ea0adb2a0eb538e2253 : SUCCESS in 2m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7d8b5a3b5c5b4c72aa93b3199c7a5df3 : FAILURE in 4m 05s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e42fb483db9f46629a84c3cf803dfdb7 : SUCCESS in 9m 48s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/23f71a20f5fd4a3e9b40daa3cd062e77 : SUCCESS in 16m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/04805510e27d46eeb592b22e0fadbc03 : SUCCESS in 7m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8d8392844d684542a09ee22d88283c8f : SUCCESS in 9m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5fa3256e059341c2a846b7e74ca06691 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e6fd409bf3b9438a974776b8b7e8125c : SUCCESS in 2m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3bd1b1fbd2f400193f714564d72796c : SUCCESS in 29m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5f681d6f4cc346f48f708cbf91a26735 : SUCCESS in 4m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d717a6ba3af94a0db31d73053b162d8d : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/27d95827d2764c4a8fd87c65327e1c8d : SUCCESS in 5m 55s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ae0d7cea5a234310aba9fc8fae02ee0b : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85cbb0e5b0664ec9b162141b85019478 : SUCCESS in 36m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25e7f037d8e243b7b48999d4c4ba3962 : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0cd052f446f769529323a7efc7e7f3f3cec82cb4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-03 15:14:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57b070d38e1c087d03c5b3df84246f704955163e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 15:16:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"923ab266ff0234223debbdf55bb088ca1558b33d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 16:47:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16cecced77164a0eb5efb368d666787a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c0d7bd9aa5e4637af834a80100f3e07 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92d8442623974010b59e11f23833dd81 : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2dd9b67415a14621ad0dd2be4ef80c7d : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca5c616cd4db4509bd1117b537442d89 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79d60449752a4fa187e6d6899192ba62 : SUCCESS in 5m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5827af4ed6c54865bde28f635c0cfde3 : SUCCESS in 16m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4cf434ad698940daaac97d721a4db187 : SUCCESS in 11m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/184b93040e7247af94c79640d6100633 : SUCCESS in 17m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9b00e1dcb4f54798b122bd7203970718 : SUCCESS in 8m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3f271f37b4784b16885c850655e66d84 : SUCCESS in 7m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/00738d2325f34201b4852f625d43901f : SUCCESS in 4m 22s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1c2458aeb4184be485f319d7a520ad4d : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cac4339bc2294bab9cfecfff2bee9fad : SUCCESS in 7m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/68ffc448f2b04065a860789e9d522601 : SUCCESS in 15m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/419dcd6b66844eb8a422c9101b01d357 : SUCCESS in 6m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/788d443d902a45a7b1c104e29e356cea : SUCCESS in 9m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fb7231736f624c69a576c5d8ae2b212c : SUCCESS in 2m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fa527128288f47839ee1f20d6ca42279 : SUCCESS in 10m 08s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/06d2c7982963480fa9241bed1955341e : SUCCESS in 16m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e782adc8054443c181b8123c4030c1c5 : SUCCESS in 7m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8e693c9bf0e243788eb1cfdfd7914154 : SUCCESS in 13m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c57dfde148c34d78bb6ccce33926334e : SUCCESS in 6m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c25f95db560d4e0291a6990905c19961 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c5b75a71cdc43959a728d1c5ffc3d6c : SUCCESS in 32m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fb67a12e5ec2428caff9677153b1bcb5 : SUCCESS in 6m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ec3bbb12eb634488aebb06eeba59b16b : SUCCESS in 4m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ab0533812f9c4f8e8e468b57d140ecf8 : SUCCESS in 18m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/574c425ddad542d1ad5b4c6a80917856 : SUCCESS in 1h 03m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/000710fc2f304ca49e6a1c9218c07ef6 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"219939ee9d18c76aec14a5e5182dde48269d7e02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 16:47:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"414ad2ac0867ee6cafd26d4240304ca379614d08","revisions":{"414ad2ac0867ee6cafd26d4240304ca379614d08":{"kind":"REWORK","_number":1,"created":"2026-08-03 12:53:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/999599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/999599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/999599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/999599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/999599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/999599/1"}}},"commit":{"parents":[{"commit":"3ef87c60c39eef0ff2dc6456c3a233b9b8b7b638","subject":"Merge \"update constraint for oslo.service to new release 4.1.2\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ef87c60c39eef0ff2dc6456c3a233b9b8b7b638"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-03 12:43:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-03 12:43:05.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 5.4.1","message":"update constraint for python-manilaclient to new release 5.4.1\n\nmeta: version: 5.4.1\nmeta: team: manila\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: I03ff2fa71a65fad888663afe67323cc5270913f4\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia9940619f6e6654d8dccac6f3be144474c8a2504\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/414ad2ac0867ee6cafd26d4240304ca379614d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/414ad2ac0867ee6cafd26d4240304ca379614d08"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
