)]}'
{"id":"openstack%2Frequirements~999719","triplet_id":"openstack%2Frequirements~master~Id07664d05089e201ecf2d8f41778f8bc9be882ad","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id07664d05089e201ecf2d8f41778f8bc9be882ad","subject":"update constraint for os-service-types to new release 1.9.0","status":"MERGED","created":"2026-08-04 13:28:16.000000000","updated":"2026-08-11 17:18:29.000000000","submitted":"2026-08-11 17:17:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999719-new-release","meta_rev_id":"8dacc1f1fcd681aae0588612e354252aaa4387f8","_number":999719,"virtual_id_number":999719,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-11 17:17:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-08-07 13:10:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-08-07 13:10:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-04 14:38:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-07 13:10:24.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-08-11 14:11:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"911d3d6e228d49f81d185209eb71f2e03b33a169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-04 13:28:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8c302a4ec192f039ab7bfad6ff2b42226b5821a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 14:38:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3df15e145aa244b49c7d37d3bff2277c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8bbf969b9e49f086c965d6f0d5fec2 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d782b6ff5f45f59d120ccbff564706 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7535edf4a72e4376a8246ca5c5e90b6b : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56f42d33bcd649a6b8b7cc8d1ac7aafc : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54d211740068497d8f1bf2daf033ec97 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29d5a1a9a15844f4b294d3c0f7a55b20 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6c6adbb6433450f914c5a6ae49b1b8b : SUCCESS in 2m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c078893a85bf414aa1d689e0e5708684 : SUCCESS in 7m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/32ef30dde6ec4f518477343ea329950c : SUCCESS in 7m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/811cde4449a54492ba50708e5a586418 : SUCCESS in 5m 57s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/e0f81a1afc6c4ba1b2fef0f355e568ba : SUCCESS in 3m 46s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/dec52d0173ab4ae488b5f0ebd893d8c4 : SUCCESS in 7m 17s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/2947ee0f036849a8bf27984179f2ed3e : SUCCESS in 5m 13s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/11194269976a48a18b8418f590897e98 : SUCCESS in 7m 30s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/75f6c45313264062ad9c5c27aca7b3ac : SUCCESS in 4m 34s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/b19ed2d2ba744466808370230c2c8acc : SUCCESS in 4m 35s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e5d08364d06b48298810be4744607f75 : SUCCESS in 3m 24s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/1c6413fab82e4dadacb410179def2e8c : SUCCESS in 17m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46be65b52cab44ab907cf99e71c09138 : SUCCESS in 6m 43s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/cd29eb76c8794309879b93aa0e75ad25 : SUCCESS in 6m 15s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0c1ad5db333f4647b27002943dae2192 : SUCCESS in 13m 19s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/8dd3eb0a55c84cb69dae0871ab78609a : SUCCESS in 2m 32s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/34b86a3ce6c84bc9803d7bec8f85eaf5 : SUCCESS in 4m 25s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/84ca16b17da24ac1994a8d71458a0fcb : SUCCESS in 3m 59s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/922bd6bfc89840d48a10b44ee24d230a : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/0ca3d71f1a54469b8a488f2232f91f5b : SUCCESS in 11m 33s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/225a454d7da04acb9d842af960b686de : SUCCESS in 28m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d6ed85cfadcb42ef9b2ec7e81a8d2d6d : SUCCESS in 8m 37s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/f4253712293545abbcc198c6d1c7a4f1 : SUCCESS in 12m 45s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/0c89a87b1ead416299fac9bcf01ac577 : SUCCESS in 2m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/67ef7dac939a4b8eac762a7ecf4b20d5 : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d218b660687e4159b4b37105dd68a448 : SUCCESS in 28m 27s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/c6a6eb82c35f4da3b84308a76181854d : SUCCESS in 7m 05s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/b0c62ceb10484860b59d179b9d5bacaa : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/528e5dbdde174e928c13ba0426f13802 : SUCCESS in 7m 49s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a5f2271d67884dcba48767aaf8a76a6b : SUCCESS in 14m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/419e54214b7d4a6ca16eb9bb9ffeb92c : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11ffcdd9abbb4ce3831201e84ecd437a : SUCCESS in 48m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3addd0710fd4013a1809ddbeacdaa3d : SUCCESS in 1h 06m 52s","accounts_in_message":[],"_revision_number":1},{"id":"da10c18ce355cbc92a8fbd68577fcfd3a8fe4178","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-07 13:10:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b562b91d3347e89bf87f44ffef1dbec3f6aceaf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 13:11:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd7a2ccc75dc1d0591951923ed9ffc0c6ea1cd58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 14:04:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d96c4ce6363f4b71989d5643566f07b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46cc8f20d184b508ede3e65ca652ef2 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b50158d2e30431ea063b78e58642e6a : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c825b8fe64354a51bc3f0560bbd87402 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5fa07cac24d4f328e8c33ae2eb38b31 : SUCCESS in 3m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfa04d32fa3d45b1a89cbf75317cb9cb : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fa47c2c264924aa09f997f67d1826255 : SUCCESS in 4m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/949a1e842ce842b38538621420acce25 : SUCCESS in 3m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2ffb5108036f44f5866f621d4c5be81f : SUCCESS in 8m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/70d3cdbf046749b68122b2db35f29b9a : SUCCESS in 5m 38s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e6356eb48a5f40888e1c4585f805daea : SUCCESS in 10m 34s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/911ed1399e7047e990b9b9c8d5406f74 : SUCCESS in 6m 16s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/066927c383bd492aad50c5a4fedcd216 : SUCCESS in 4m 42s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/25ac5915867c46a0a3528b72a86d4b80 : SUCCESS in 5m 52s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a68e322b516e4dd6b1acb1724cc8e841 : SUCCESS in 17m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a2f5acc059c4c5c91338368b39e9a83 : SUCCESS in 6m 10s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/be12fa20df8e4aa187eeef2cae18c815 : SUCCESS in 5m 16s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/6f2ce4a29648412899a09e16439e779e : SUCCESS in 10m 18s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/bc82dca1eab848a5aa150126136fdd76 : SUCCESS in 2m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/9c06a25e7f064bd69e2fc88b2976c81d : SUCCESS in 10m 04s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/9e877f8c11bb453e861c8621af0ebac7 : SUCCESS in 23m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ad2d2208feae4f96a29651c338c41bc4 : SUCCESS in 7m 17s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/18e4d519a9614912a7ba907c86a16186 : SUCCESS in 12m 58s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/044f2e6a87034bb193df7f873bad545f : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b24905e3569b4b0588b028b4e61e035a : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c54b0d2984ba431e8fad02c9c942195c : SUCCESS in 24m 58s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4873b6689aa44f3c8bb902e67bc01852 : SUCCESS in 5m 10s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/46eb7cf951fb42e790029dbb06cf5f26 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c24ac6d5dda244f39895b24f8a7af9cc : SUCCESS in 6m 20s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/e56ad31746d349f6b42aadf2417571a5 : SUCCESS in 9m 02s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5e89503cfbc34f2aaf6acf7a5b5e42c4 : SUCCESS in 3m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b9834ebce5c46f99bba0de9de6a36f5 : FAILURE in 34m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0dada32408af42d8acd2fcb6e6178edc : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1c3996c451e241a79698081687a7c0de7452883a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-11 14:11:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d432793e0e3c416158401880d31a4a23c4f55514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 15:54:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7198138afdb149afa42efaf1b3cc44dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e5c4ce74c74815b5b9f7886c4894b2 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/059d41f704cd4982a45abccec3947aaa : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e81dcff686e4a4099b73a43ccfe8809 : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eafe1d1ea74d4e2f84fd794bb3a07d10 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d4e42fdf0796459db3430c6736322e11 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af6e3f01cfa94827808d42154eb19529 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7b3fe146eaa45ee829706955c6bdcdf : SUCCESS in 4m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/86b1600b4e834eb09b887870432d45a8 : SUCCESS in 3m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5e437e5c5d134181837ba2cb091891a5 : SUCCESS in 6m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c4cf8639a4d243298eca6bf4581b90c1 : SUCCESS in 6m 13s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/cdcd5ce770554d22bce2b1fcf90d39c0 : SUCCESS in 4m 32s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/a6ba982f803e4fe69c1a1b23adfdb562 : SUCCESS in 4m 55s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/568c799ce8b1442f813a8b0122c8e6a6 : SUCCESS in 5m 58s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/8f4ae439f0964e40a79ef2204ad17607 : SUCCESS in 12m 21s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/54dcec276e5f41789633da7d8b359e9f : SUCCESS in 5m 02s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/ff39e34039d1421fb39e0236411d17b4 : SUCCESS in 3m 38s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9cd877c9849145438aed28ae2eb7c82f : SUCCESS in 6m 06s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/eb7c2a64e0f34b4fb44788922537ebc8 : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54579aa50ef54ac99b21903e43ead5af : SUCCESS in 4m 55s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/e1f017a7f87a49b280d8f75b113a500e : SUCCESS in 7m 44s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/78ba59156ec84fd08279076ec425503b : SUCCESS in 10m 51s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c73b149dac3848378d697e242c700d42 : SUCCESS in 3m 19s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/b9c54846c440402d99051cbc9cc165e4 : SUCCESS in 5m 25s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/3f52f8e70aab4d3a85f6696812dcfb4e : SUCCESS in 4m 47s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/40fdb0c912854baba1c6330f751fa1d1 : SUCCESS in 4m 31s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a656fbb631294233b5614dc855fdac35 : SUCCESS in 11m 13s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0744b74791d54ec191bb5cdcd2bd39c6 : SUCCESS in 17m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1f242e0040ef4f9a8d19778f1e8af2a7 : SUCCESS in 6m 59s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/ea393683c0eb484e9f39b72866820fbe : SUCCESS in 13m 38s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/e0f248fdd0884ed0b3a91701310371c6 : SUCCESS in 3m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b9a4eb61bece4665a187ab41a7b5b3ff : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/904a279c876442408e0e7dc64fa76250 : SUCCESS in 20m 42s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/447f3a29d66244108ca62db7d38994c6 : SUCCESS in 5m 39s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/2763b1c0485c49aabedb53b7553f01fc : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d53ed02262024dddaed13059bc6815bd : SUCCESS in 7m 15s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/b1bdb9e834ee49d0b69b61747cfe123e : SUCCESS in 8m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/de8d762d086b484eb47231362b516315 : SUCCESS in 5m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b45baf8eb6ab4f9784f6089008964f0b : SUCCESS in 33m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b90b73eee3284d2186ecbeae70070953 : SUCCESS in 1h 27m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2d784efad8333b7743f4023edb02e6eeaf999a24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 15:55:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b679a85cf7918af2b939854798b3e573c6f6ba4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 17:17:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d40a1527dbf4f16aab30a9b7d098739\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d50987a85c4e3389936f45a229869f : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf58721ce1f4135a2b08134b4a4ac3b : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aacf882a18424deead846d5249aba200 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efe44aed448340ca9c893f58c8469b73 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2a775e448204a3bb17408157969ab3f : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7b1739f0896420dbeccab5ec8593a8b : SUCCESS in 4m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8a56d0f7f069474abc501eb21f9f10e8 : SUCCESS in 7m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1b3479e2ea83441f94cac206b9310a80 : SUCCESS in 6m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/57461bc2356940798eb4260ceb5a766a : SUCCESS in 5m 44s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/9a9aa41ba21c4471ac8bb362f2f90d9e : SUCCESS in 7m 44s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/20005928acb847998ff352e57ecb47e8 : SUCCESS in 4m 39s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/22d1f231bc34497b9a93939e2d10d97e : SUCCESS in 5m 29s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/a7884f5b71a64a18b782474eabb46904 : SUCCESS in 6m 06s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/7260508b43ad46af8fd2da653ba3dbd5 : SUCCESS in 17m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e56b4a86deee4356b5cddea4b6815c3f : SUCCESS in 6m 47s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/39e6d9eb66184a56b618a733b67f890e : SUCCESS in 4m 21s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4df74d1035f9433ca66ba156ff9c96d8 : SUCCESS in 9m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/6974573fbad14b2582868215e4a3465f : SUCCESS in 3m 40s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f456a628eb414e20a7fafef9769a2b6e : SUCCESS in 6m 15s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/a765a680c19540be992dc4ee8b3782f2 : SUCCESS in 22m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/99c6125a29f2476ba5f868f80a5b930d : SUCCESS in 6m 54s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/38223ac1d4cf4d77b184be268bcdffe3 : SUCCESS in 9m 00s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/59b4bdec572d4d68ade5dd228074c278 : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ff396b733aa4b1683f79c236f65f274 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9280c7c166174337ad108e7bcc829a18 : SUCCESS in 27m 08s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/013a42a4720247c2b4dd36314354e859 : SUCCESS in 5m 21s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/7806389ed3784c298da038dd7e17ee3c : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/acfa0ec78b204752b954e50f424faff2 : SUCCESS in 3m 48s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a47ec88db44146b6b4f69a7bb8c5e9fe : SUCCESS in 8m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/eaa3df5d94db47dca8803f3222d10517 : SUCCESS in 3m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ad6dd389198424fbdfb4da17f9d2b0a : SUCCESS in 1h 04m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d09e7ede4354c9e892724459d380c6e : SUCCESS in 1h 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"498e6bdf390c2f17ed0df4a3f82383537977bae3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 17:17:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8dacc1f1fcd681aae0588612e354252aaa4387f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 17:18:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1c0b7ccd2784bc6a8ea274c9bfade37\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43dbc57e930843f39f9824c61d1dadfc : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b928f1a68fc7f110fd4a7efbaf908b0e69017304","revisions":{"b928f1a68fc7f110fd4a7efbaf908b0e69017304":{"kind":"REWORK","_number":1,"created":"2026-08-04 13:28:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/999719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/999719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/999719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/999719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/999719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/999719/1"}}},"commit":{"parents":[{"commit":"22c7e800bda7e90d83ab215df50a610477e623d0","subject":"Merge \"update constraint for sushy to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22c7e800bda7e90d83ab215df50a610477e623d0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-04 13:18:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-04 13:18:13.000000000","tz":0},"subject":"update constraint for os-service-types to new release 1.9.0","message":"update constraint for os-service-types to new release 1.9.0\n\nmeta: version: 1.9.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I97562506de3debac30abedc49397d59ab19afb42\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id07664d05089e201ecf2d8f41778f8bc9be882ad\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b928f1a68fc7f110fd4a7efbaf908b0e69017304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b928f1a68fc7f110fd4a7efbaf908b0e69017304"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
