)]}'
{"id":"openstack%2Frpm-packaging-tools~674794","triplet_id":"openstack%2Frpm-packaging-tools~master~Iba609c18e173301ca8c4d4c584431a98f332dad3","project":"openstack/rpm-packaging-tools","branch":"master","topic":"improve","hashtags":[],"change_id":"Iba609c18e173301ca8c4d4c584431a98f332dad3","subject":"Use python3 in shebang","status":"MERGED","created":"2019-08-06 12:14:17.000000000","updated":"2019-08-08 15:46:20.000000000","submitted":"2019-08-08 15:46:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674794-1565279180932-484ec8b8","meta_rev_id":"fa3b3d4a6352254b7661f2e267feffb9484076a0","_number":674794,"virtual_id_number":674794,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-08-08 15:46:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-08-08 14:55:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"date":"2019-08-08 06:49:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-08 15:34:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 06:49:24.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-08-08 14:55:50.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-08-08 15:46:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df647c7b5f54743f5f2563e886c074e080f2e303","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-06 12:14:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d96c570f0b594453a175467b134109e8701d1103","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 12:25:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/94/674794/1/check/openstack-tox-pep8/88c3857/ : SUCCESS in 3m 35s\n- rpm-packaging-status https://logs.opendev.org/94/674794/1/check/rpm-packaging-status/2d585c5/html/ : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1a7ce52ab27ce6b70a5dc2ad78e254ca02024bbb","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-08 14:55:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"133c4b9db99e1d4557a7430ffbe5e5c2030404ff","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-08 15:34:02.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9c95c43783e2b9b7d52361fc4147ec49463261d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:34:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b9e6895e1e35bcead7ba9aa7f617872103eb80a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:46:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a552f8c3aa9a514c2c3ae6a6cc0417983e2c6afa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:46:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/94/674794/1/gate/openstack-tox-pep8/861fbcf/ : SUCCESS in 3m 39s\n- rpm-packaging-status https://logs.opendev.org/94/674794/1/gate/rpm-packaging-status/77ebaeb/html/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59c9ae3dc5be1ec410a8cb0ba4063ec9fbcc5628","revisions":{"59c9ae3dc5be1ec410a8cb0ba4063ec9fbcc5628":{"kind":"REWORK","_number":1,"created":"2019-08-06 12:14:17.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/94/674794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging-tools","ref":"refs/changes/94/674794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging-tools refs/changes/94/674794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging-tools refs/changes/94/674794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging-tools refs/changes/94/674794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging-tools refs/changes/94/674794/1"}}},"commit":{"parents":[{"commit":"c33ad8f7852a5a1ef0bb527e1de182f6776d8992","subject":"Follow opendev rebranding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging-tools/commit/c33ad8f7852a5a1ef0bb527e1de182f6776d8992"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-08-06 11:34:44.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-08-06 11:34:44.000000000","tz":120},"subject":"Use python3 in shebang","message":"Use python3 in shebang\n\nGet rid of python2\n\nChange-Id: Iba609c18e173301ca8c4d4c584431a98f332dad3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging-tools/commit/59c9ae3dc5be1ec410a8cb0ba4063ec9fbcc5628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging-tools/commit/59c9ae3dc5be1ec410a8cb0ba4063ec9fbcc5628"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
