)]}'
{"id":"openstack%2Frpm-packaging~388461","triplet_id":"openstack%2Frpm-packaging~master~I5825ca42cbff78cabbd63f22192ecc869cb88fc9","project":"openstack/rpm-packaging","branch":"master","topic":"update-oslo.middleware","hashtags":[],"change_id":"I5825ca42cbff78cabbd63f22192ecc869cb88fc9","subject":"Update oslo.middleware to 3.20.0","status":"MERGED","created":"2016-10-19 06:46:26.000000000","updated":"2016-10-20 07:46:27.000000000","submitted":"2016-10-20 07:27:22.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92dc0aa3f5e20ac3b59c949da53e48dccf606172","_number":388461,"virtual_id_number":388461,"owner":{"_account_id":22266,"name":"Tony Xu","email":"hhktony@gmail.com","username":"xutao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2016-10-19 06:49:37.000000000","_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-10-19 07:08:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2016-10-20 07:27:22.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-10-20 07:25:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-10-20 05:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-10-20 07:25:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-10-19 06:49:37.000000000","updated_by":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-10-19 07:08:57.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-10-20 05:45:56.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2016-10-20 07:25:23.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2016-10-20 07:27:22.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"cac01bba89cdf9c75f82f3383ee1ff9bd6166721","author":{"_account_id":22266,"name":"Tony Xu","email":"hhktony@gmail.com","username":"xutao"},"date":"2016-10-19 06:46:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9991351459634edcb48b6d37e18b957569ea6fff","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2016-10-19 06:46:32.000000000","message":"Patch Set 1:\n\nBuild queued (rpm_test pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":1},{"id":"f390fcfa580e48ab724b9d865bc184453f376fde","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2016-10-19 06:49:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm_test pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/1325/ : SUCCESS in 3m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ced2422ba478a267033e0873c450b38bb763ea93","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-19 06:52:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/61/388461/1/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/8b2a3dd/ : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d25fcd998c5166f8730722027d9d729fdd6c592f","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2016-10-19 07:08:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-9c1ad49a24183dadea0cd82cc7bf43050afe4170 : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bd10997835b502a61471e17e2cdd3dda342b2b1f","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2016-10-20 05:45:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"48349372f816ca44aa4c3f73f2d81201ddaa888a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2016-10-20 07:25:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e33e93ad30431442c10120eaf68d96e993f2919","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-20 07:25:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"b67a2a24c7504f1c6195cdb063da60527367487f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-20 07:27:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/61/388461/1/gate/gate-rpm-packaging-tox-lint-ubuntu-xenial/ffee010/ : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cd3ca33e6f1686a75be3dabe12985d81f93b11bd","date":"2016-10-20 07:27:23.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"cd5c769d8455d8e7217bb2a740f1f19457ad9f22","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2016-10-20 07:27:28.000000000","message":"Patch Set 1:\n\nBuild queued (rpm_publish pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":1},{"id":"ab103874cc31751122fb46ec3669b121db011668","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2016-10-20 07:46:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (rpm_publish pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/1365/ : SUCCESS in 18m 56s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c1ad49a24183dadea0cd82cc7bf43050afe4170","revisions":{"9c1ad49a24183dadea0cd82cc7bf43050afe4170":{"kind":"REWORK","_number":1,"created":"2016-10-19 06:46:26.000000000","uploader":{"_account_id":22266,"name":"Tony Xu","email":"hhktony@gmail.com","username":"xutao"},"ref":"refs/changes/61/388461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/61/388461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/61/388461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/61/388461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/61/388461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/61/388461/1"}}},"commit":{"parents":[{"commit":"906bad360db1fbaedaf9cec7722648871ac2c3e7","subject":"Merge \"Add initial spec for yaql\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/906bad360db1fbaedaf9cec7722648871ac2c3e7"}]}],"author":{"name":"Tony Xu","email":"hhktony@gmail.com","date":"2016-10-19 06:45:38.000000000","tz":480},"committer":{"name":"Tony Xu","email":"hhktony@gmail.com","date":"2016-10-19 06:45:38.000000000","tz":480},"subject":"Update oslo.middleware to 3.20.0","message":"Update oslo.middleware to 3.20.0\n\nChange-Id: I5825ca42cbff78cabbd63f22192ecc869cb88fc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9c1ad49a24183dadea0cd82cc7bf43050afe4170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9c1ad49a24183dadea0cd82cc7bf43050afe4170"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"906bad360db1fbaedaf9cec7722648871ac2c3e7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
