)]}'
{"id":"openstack%2Frpm-packaging~388474","triplet_id":"openstack%2Frpm-packaging~master~I233935002fbff30ae041ad639e3fb00de8a38a1f","project":"openstack/rpm-packaging","branch":"master","topic":"update-oslo.rootwrap","hashtags":[],"change_id":"I233935002fbff30ae041ad639e3fb00de8a38a1f","subject":"Update oslo.rootwrap to 5.2.0","status":"MERGED","created":"2016-10-19 07:00:56.000000000","updated":"2016-10-20 05:44:25.000000000","submitted":"2016-10-20 05:42:52.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b1d51de52e67a7c53d376a2f3067c71140d9219b","_number":388474,"virtual_id_number":388474,"owner":{"_account_id":22266,"name":"Tony Xu","email":"hhktony@gmail.com","username":"xutao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2016-10-19 07:04:05.000000000","_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-10-19 07:20:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2016-10-20 05:42:52.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-10-19 20:50:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-10-20 05:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-10-20 05:41:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-10-19 07:04:05.000000000","updated_by":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-10-19 07:20:34.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-10-19 20:50:45.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2016-10-20 05:41:26.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2016-10-20 05:42:52.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"1b095c7557930fb0a009b690eed6b40742da45e9","author":{"_account_id":22266,"name":"Tony Xu","email":"hhktony@gmail.com","username":"xutao"},"date":"2016-10-19 07:00:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"062d18e5a266db9f9e964dad5d3cd706f6542c66","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2016-10-19 07:01:03.000000000","message":"Patch Set 1:\n\nBuild queued (rpm_test pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":1},{"id":"c94886fbf24f3e35547044e22c59f6f9c570d028","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-19 07:03:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/74/388474/1/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/9483901/ : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3874d8332c3e6628b5f169049d60b728c24cbd0f","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2016-10-19 07:04:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm_test pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/1329/ : SUCCESS in 2m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"567ea6387ba006861479862f51428ded00a5b473","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2016-10-19 07:20:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-a46aa11d167bf37619c46e8d38acb7aaede41b2f : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"eb38be606d99c9579fe5fb184760675485431297","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2016-10-19 20:50:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa93893c28c2db28d17e229149497bdb0e06d991","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2016-10-20 05:41:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"584a61d9816de3307b66af50ac9f72e7234ec3e5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-20 05:41:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"086d6d4c5c2e0d46416fb6e902e93f64d02757ef","date":"2016-10-20 05:42:52.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"3856064e518e8f9dbe367da66a774d3d069c08b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-20 05:42:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/74/388474/1/gate/gate-rpm-packaging-tox-lint-ubuntu-xenial/f5fbc23/ : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8dd3962a007a09780b6aaa16c31f61f312107b1c","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2016-10-20 05:42:58.000000000","message":"Patch Set 1:\n\nBuild queued (rpm_publish pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":1},{"id":"3976cfd782dc770e44dd82e6c4cd906ff55ec014","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2016-10-20 05:44:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (rpm_publish pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/1355/ : FAILURE in 1m 23s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a46aa11d167bf37619c46e8d38acb7aaede41b2f","revisions":{"a46aa11d167bf37619c46e8d38acb7aaede41b2f":{"kind":"REWORK","_number":1,"created":"2016-10-19 07:00:56.000000000","uploader":{"_account_id":22266,"name":"Tony Xu","email":"hhktony@gmail.com","username":"xutao"},"ref":"refs/changes/74/388474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/74/388474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/74/388474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/74/388474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/74/388474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/74/388474/1"}}},"commit":{"parents":[{"commit":"906bad360db1fbaedaf9cec7722648871ac2c3e7","subject":"Merge \"Add initial spec for yaql\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/906bad360db1fbaedaf9cec7722648871ac2c3e7"}]}],"author":{"name":"Tony Xu","email":"hhktony@gmail.com","date":"2016-10-19 07:00:33.000000000","tz":480},"committer":{"name":"Tony Xu","email":"hhktony@gmail.com","date":"2016-10-19 07:00:33.000000000","tz":480},"subject":"Update oslo.rootwrap to 5.2.0","message":"Update oslo.rootwrap to 5.2.0\n\nChange-Id: I233935002fbff30ae041ad639e3fb00de8a38a1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a46aa11d167bf37619c46e8d38acb7aaede41b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a46aa11d167bf37619c46e8d38acb7aaede41b2f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"906bad360db1fbaedaf9cec7722648871ac2c3e7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
