)]}'
{"id":"openstack%2Frpm-packaging~422009","triplet_id":"openstack%2Frpm-packaging~master~I48757331487eb2197183a4e9e2c661f0249865cd","project":"openstack/rpm-packaging","branch":"master","topic":"oslo-messaging","hashtags":[],"change_id":"I48757331487eb2197183a4e9e2c661f0249865cd","subject":"Update oslo.messaging to 5.17.0","status":"MERGED","created":"2017-01-18 15:18:54.000000000","updated":"2017-01-26 17:20:25.000000000","submitted":"2017-01-26 17:17:44.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"494938815bd9584440b0e0687f2f50386c752ad6","_number":422009,"virtual_id_number":422009,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2017-01-24 23:37:36.000000000","_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-01-25 00:19:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10384,"name":"Haïkel Guémar","email":"hguemar@fedoraproject.org","username":"hguemar"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2017-01-26 17:17:44.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7613,"name":"Igor Yozhikov","email":"medput@gmail.com","username":"iyozhikov"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10384,"name":"Haïkel Guémar","email":"hguemar@fedoraproject.org","username":"hguemar"},{"value":2,"date":"2017-01-25 00:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2017-01-25 11:16:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2017-01-26 11:14:21.000000000","_account_id":7613,"name":"Igor Yozhikov","email":"medput@gmail.com","username":"iyozhikov"},{"value":2,"date":"2017-01-26 17:15:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-01-26 13:04:19.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10384,"name":"Haïkel Guémar","email":"hguemar@fedoraproject.org","username":"hguemar"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7613,"name":"Igor Yozhikov","email":"medput@gmail.com","username":"iyozhikov"},{"value":1,"date":"2017-01-26 17:15:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7613,"name":"Igor Yozhikov","email":"medput@gmail.com","username":"iyozhikov"},{"_account_id":10384,"name":"Haïkel Guémar","email":"hguemar@fedoraproject.org","username":"hguemar"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-24 23:24:35.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-24 23:37:36.000000000","updated_by":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-25 00:19:22.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-25 00:32:14.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2017-01-25 11:16:32.000000000","updated_by":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"reviewer":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"state":"REVIEWER"},{"updated":"2017-01-26 11:14:21.000000000","updated_by":{"_account_id":7613,"name":"Igor Yozhikov","email":"medput@gmail.com","username":"iyozhikov"},"reviewer":{"_account_id":7613,"name":"Igor Yozhikov","email":"medput@gmail.com","username":"iyozhikov"},"state":"REVIEWER"},{"updated":"2017-01-26 13:04:19.000000000","updated_by":{"_account_id":10384,"name":"Haïkel Guémar","email":"hguemar@fedoraproject.org","username":"hguemar"},"reviewer":{"_account_id":10384,"name":"Haïkel Guémar","email":"hguemar@fedoraproject.org","username":"hguemar"},"state":"REVIEWER"},{"updated":"2017-01-26 17:15:11.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2017-01-26 17:17:44.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"89e97921f822124cb02f6ee4c1c28415a636767c","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2017-01-18 15:18:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"853e998082de8bbcbbf1ffe14bffa3e582321a8b","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-01-18 15:19:01.000000000","message":"Patch Set 1:\n\nBuild queued (rpm_test pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":1},{"id":"2a104e69de521641c0968147cfcef8d062690883","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-01-18 15:20:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm_test pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/2400/ : FAILURE in 1m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"743bfddfbc5a9c29dbd2ba9050728a3a82f59d9c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-01-18 15:24:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/814/ : FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"54c3fc3520c23dac4d107d6cb9341a6810214a74","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-18 15:29:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/09/422009/1/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/c0c8577/ : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3d064d4fec253aea6f20b624ee5be23e57019559","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2017-01-18 15:31:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-76bf78d63190daab31c7c7812820b1542f4157c2 : FAILURE in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e6faf662658ab79bd857622f616c5857da6457c0","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2017-01-20 21:33:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d00a0adab45fba078dd706e093f43832fb22a194","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-01-20 21:33:28.000000000","message":"Patch Set 1: -Verified\n\nBuild queued (rpm_test pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":1},{"id":"3a9b79ac7d71e5a1e2db080029e85d8a3776b494","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-01-20 21:35:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm_test pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/2534/ : FAILURE in 1m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abfd232f895699bca6d47505c6028504b633c87d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-20 21:35:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/09/422009/1/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/0a2a43d/ : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1},{"id":"6e08a9d65ed486dd0d7b0dc0bd6fada2ff23782f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-01-20 21:38:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/912/ : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7bd56e5fe7acb73189cd321868877c6bd97545e4","author":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"date":"2017-01-23 09:19:08.000000000","message":"Patch Set 1: Code-Review-1\n\nThe failing test point to some problem with zmq import, that I do not see that in the requires.","accounts_in_message":[],"_revision_number":1},{"id":"2745a2b5f01d7c57e922e37bd46303d86898382c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2017-01-24 23:19:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a969d359c3f5a2455cf6de4f471f6217d827d1aa","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 23:19:32.000000000","message":"Patch Set 2:\n\nBuild queued (rpm_test pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":2},{"id":"01dc17f5597ace0af6dd4f4408cb6fc578d348ff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 23:20:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/09/422009/2/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/533b1f3/ : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2},{"id":"87ddb62bd16e27d6b7b2c770d6166cec1dcfe6d1","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 23:21:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (rpm_test pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/2573/ : FAILURE in 2m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac492205d116386c42599e10f1dc2e1c1f90282c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-01-24 23:24:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/955/ : FAILURE in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e58c97c1c3834c3f9ea76c7a2d27a20fe501c311","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 23:26:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-3b7f9d08c540b34989e96985633916c90d80a807 : FAILURE in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b6e513d7671cad7fe83e3bf3cc00bd080fd0a1d9","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2017-01-24 23:33:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"454cedeac09e49fc2565dc1798fb7b65c40c33f6","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 23:33:34.000000000","message":"Patch Set 3:\n\nBuild queued (rpm_test pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":3},{"id":"4e7bae2349faaf27d04c1e0c0455989d32a43958","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 23:37:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/09/422009/3/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/abc76a0/ : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3},{"id":"4872281abf362e13da3568ed5eecb900f5a0e6a9","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 23:37:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (rpm_test pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/2579/ : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7da15840b933abb7d63e32dbf0c6fe5005d34d38","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-01-24 23:38:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/961/ : FAILURE in 4m 39s","accounts_in_message":[],"_revision_number":3},{"id":"827b3801eb4a730983a6e147b8ab0714effcc14f","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2017-01-25 00:19:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-a8e0a57829a64cff49b3c9d8fb2288ab798ee682 : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":3},{"id":"d39fe54a4a40ff2e5baec8adf6688ee430246caa","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2017-01-25 00:32:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"106fc0b582bb69259fb6756a109a6f78791730e5","author":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"date":"2017-01-25 11:16:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4fceff5c051ef6e742f5ee905e7229f0118fa532","author":{"_account_id":7613,"name":"Igor Yozhikov","email":"medput@gmail.com","username":"iyozhikov"},"date":"2017-01-26 11:14:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4f839f2b33bf22094b23c91c6669a263a1aa03a2","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2017-01-26 13:04:58.000000000","message":"Patch Set 3:\n\nWe need a higher version of python-kafka in the RDO deps repo.","accounts_in_message":[],"_revision_number":3},{"id":"2b45155e120e642f65cb598e2c0081e155d1260b","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2017-01-26 17:06:12.000000000","message":"Patch Set 3:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":3},{"id":"e9f68a7140e620dbf58ae5ca78329eb0e89af64b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-01-26 17:13:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/1013/ : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":3},{"id":"7f8e1a38711a0080cc680556a326b82e21a648c8","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2017-01-26 17:15:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"54f5cd78ecb39aa1a75ab4d6217f78d4e25e1d64","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 17:15:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"618f06614c0d8a17afee436d822492e3c642b26b","date":"2017-01-26 17:17:44.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"89f7a8efa38dd7a8149f6076a98a52a4f5de6b79","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 17:17:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/09/422009/3/gate/gate-rpm-packaging-tox-lint-ubuntu-xenial/365b275/ : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":3},{"id":"ed1f971146f478131113bdc13fc6975d497a69a7","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 17:17:50.000000000","message":"Patch Set 3:\n\nBuild queued (rpm_publish pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":3},{"id":"1766d0a90f0b16e267b69b0673fe3d1db7d7862a","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 17:20:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (rpm_publish pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/2651/ : SUCCESS in 2m 31s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a8e0a57829a64cff49b3c9d8fb2288ab798ee682","revisions":{"76bf78d63190daab31c7c7812820b1542f4157c2":{"kind":"REWORK","_number":1,"created":"2017-01-18 15:18:54.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/09/422009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/422009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/1"}}},"commit":{"parents":[{"commit":"21c51aac56dd4cbd7b6d708249a981d343310b53","subject":"Merge \"Update os-brick to 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/21c51aac56dd4cbd7b6d708249a981d343310b53"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2017-01-18 15:15:25.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2017-01-18 15:15:25.000000000","tz":480},"subject":"Update oslo.messaging to 5.17.0","message":"Update oslo.messaging to 5.17.0\n\nChange-Id: I48757331487eb2197183a4e9e2c661f0249865cd\nDepends-on: Ie8ea4929a9535621f4b1e2e0aa800439b170bd54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/76bf78d63190daab31c7c7812820b1542f4157c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/76bf78d63190daab31c7c7812820b1542f4157c2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"21c51aac56dd4cbd7b6d708249a981d343310b53","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3b7f9d08c540b34989e96985633916c90d80a807":{"kind":"REWORK","_number":2,"created":"2017-01-24 23:19:26.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/422009/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/422009/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/2"}}},"commit":{"parents":[{"commit":"063cacce3058422a3934f5cd2604ab4f03a71ed0","subject":"Merge \"Update os-client-config to 1.26.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/063cacce3058422a3934f5cd2604ab4f03a71ed0"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2017-01-18 15:15:25.000000000","tz":480},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2017-01-24 23:19:18.000000000","tz":60},"subject":"Update oslo.messaging to 5.17.0","message":"Update oslo.messaging to 5.17.0\n\nChange-Id: I48757331487eb2197183a4e9e2c661f0249865cd\nDepends-on: Ie8ea4929a9535621f4b1e2e0aa800439b170bd54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3b7f9d08c540b34989e96985633916c90d80a807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3b7f9d08c540b34989e96985633916c90d80a807"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"063cacce3058422a3934f5cd2604ab4f03a71ed0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a8e0a57829a64cff49b3c9d8fb2288ab798ee682":{"kind":"REWORK","_number":3,"created":"2017-01-24 23:33:27.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/422009/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/422009/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/422009/3"}}},"commit":{"parents":[{"commit":"063cacce3058422a3934f5cd2604ab4f03a71ed0","subject":"Merge \"Update os-client-config to 1.26.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/063cacce3058422a3934f5cd2604ab4f03a71ed0"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2017-01-18 15:15:25.000000000","tz":480},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2017-01-24 23:33:19.000000000","tz":60},"subject":"Update oslo.messaging to 5.17.0","message":"Update oslo.messaging to 5.17.0\n\nChange-Id: I48757331487eb2197183a4e9e2c661f0249865cd\nDepends-on: Ie8ea4929a9535621f4b1e2e0aa800439b170bd54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a8e0a57829a64cff49b3c9d8fb2288ab798ee682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a8e0a57829a64cff49b3c9d8fb2288ab798ee682"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"063cacce3058422a3934f5cd2604ab4f03a71ed0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
