)]}'
{"id":"openstack%2Frpm-packaging~442902","triplet_id":"openstack%2Frpm-packaging~master~I676931ae84a7e6b2ef538192bf10078c5dd2d21c","project":"openstack/rpm-packaging","branch":"master","topic":"oslo-versionedobjects","hashtags":[],"change_id":"I676931ae84a7e6b2ef538192bf10078c5dd2d21c","subject":"Update oslo.versionedobjects to 1.22.0","status":"MERGED","created":"2017-03-08 02:47:06.000000000","updated":"2017-03-09 21:42:11.000000000","submitted":"2017-03-09 21:40:21.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9fc6e4f2e99f81ef7a0074e1e34bba8c62428a88","_number":442902,"virtual_id_number":442902,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2017-03-08 02:50:41.000000000","_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-03-09 20:34:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2017-03-09 21:40:20.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2017-03-09 21:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-03-08 09:47:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-03-09 21:39:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-08 02:50:41.000000000","updated_by":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-03-08 09:47:23.000000000","updated_by":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"reviewer":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"state":"REVIEWER"},{"updated":"2017-03-09 20:34:02.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-03-09 21:39:06.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2017-03-09 21:40:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"8a6e96b22a934581dc305fc7f385cf57344a6aa5","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2017-03-08 02:47:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eac7bf4b0d7d8306af7783b7b30c53bf2f147831","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-03-08 02:47:12.000000000","message":"Patch Set 1:\n\nBuild queued (rpm_test pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":1},{"id":"90dfc5c96fc92213a87157f127ec971dc400366b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-08 02:48:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/02/442902/1/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/43e473d/ : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1},{"id":"661300727c382427c19ad5900e9c2c6911c291d9","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-03-08 02:50:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm_test pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/3138/ : SUCCESS in 3m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18b090dfe618a0c266fbe22dc423b9385ae8425d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-03-08 02:51:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/1506/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"399299d491b8c7db6cf7837fa759abd8c9b87abb","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2017-03-08 02:53:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-1ec0749f7500c1e4fa632cb877955b386a176fc9 : FAILURE in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f0f42ba55f4eac6838d0fed42c23b7916579d8cd","author":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"date":"2017-03-08 09:47:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"532c87f1fd4644d19941777c39bf8acfe3027ab7","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2017-03-09 19:37:57.000000000","message":"Patch Set 1:\n\nrecheck-suse","accounts_in_message":[],"_revision_number":1},{"id":"45a3d42d9498bf756d4741aacd630004ab684c65","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-09 19:41:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/02/442902/1/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/b5173e5/ : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3d2c0fd55dcf827ad166e35d336d3d344f4b3266","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-03-09 19:46:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/1559/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"98f2b352e755bb09f361351aabd1d0fdec590d09","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2017-03-09 20:34:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-1ec0749f7500c1e4fa632cb877955b386a176fc9 : SUCCESS in 22m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ff9bf5a75443446ec72954ef35ae67b28d6a52dc","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2017-03-09 21:39:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f814c16ba8d82dd063e85812504a7dcff01af89c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-09 21:39:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"5a4ef376986a6f280f5797c9e412202af72b2801","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-09 21:40:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/02/442902/1/gate/gate-rpm-packaging-tox-lint-ubuntu-xenial/986a325/ : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1},{"id":"0896d8fae5d34c227e2f1fdc2ee212d2b641f9b7","date":"2017-03-09 21:40:21.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"ab1fc67cbc018a4aae9a7429c06a3e94b19ea555","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-03-09 21:40:26.000000000","message":"Patch Set 1:\n\nBuild queued (rpm_publish pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":1},{"id":"c3f5a69eb2256e14babf0440a06cfdc7bac6b27c","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-03-09 21:42:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (rpm_publish pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/3167/ : SUCCESS in 1m 41s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ec0749f7500c1e4fa632cb877955b386a176fc9","revisions":{"1ec0749f7500c1e4fa632cb877955b386a176fc9":{"kind":"REWORK","_number":1,"created":"2017-03-08 02:47:06.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/02/442902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/02/442902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/02/442902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/02/442902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/02/442902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/02/442902/1"}}},"commit":{"parents":[{"commit":"3271c71837408370a122671a6c6520f7e8de0111","subject":"Update tooz to 1.50.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3271c71837408370a122671a6c6520f7e8de0111"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2017-03-08 02:46:06.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2017-03-08 02:46:06.000000000","tz":480},"subject":"Update oslo.versionedobjects to 1.22.0","message":"Update oslo.versionedobjects to 1.22.0\n\nChange-Id: I676931ae84a7e6b2ef538192bf10078c5dd2d21c\nDepends-on: Ie1a84aed7b378996fb5cfe77d2dd17375c6b6c83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1ec0749f7500c1e4fa632cb877955b386a176fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1ec0749f7500c1e4fa632cb877955b386a176fc9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3271c71837408370a122671a6c6520f7e8de0111","is_merged_in_target_branch":true,"change_id":"Idbee6ee69ad429d10e4b3c6850c59082eea17199","change_number":441741,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
