)]}'
{"id":"openstack%2Frpm-packaging~454089","triplet_id":"openstack%2Frpm-packaging~master~I001446da2dfb09098d55cc9bdad5b003ba01a0c5","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"I001446da2dfb09098d55cc9bdad5b003ba01a0c5","subject":"Bump openstack-macros version","status":"ABANDONED","created":"2017-04-06 10:28:25.000000000","updated":"2017-04-06 11:47:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"48586794c49e3f248b5d773a04d9cba0839b3b9c","_number":454089,"virtual_id_number":454089,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2017-04-06 11:31:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-04-06 10:31:00.000000000","_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-04-06 10:29:32.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"value":0,"date":"2017-04-06 10:32:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-04-06 10:36:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-04-06 10:29:32.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-04-06 10:31:00.000000000","updated_by":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-04-06 10:32:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-04-06 10:36:24.000000000","updated_by":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"reviewer":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"state":"REVIEWER"},{"updated":"2017-04-06 11:31:59.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d8299d60a068e52621159547718939eca5c485c","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2017-04-06 10:28:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4e13f26424c050747cb4c80f0df6d74e5a94672","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-04-06 10:28:32.000000000","message":"Patch Set 1:\n\nBuild queued (rpm_test pipeline).\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":1},{"id":"6a07e5d2a150daa5d92bc302193043b9676017b4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-06 10:29:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/89/454089/1/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/82bd68f/ : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1},{"id":"0f61d1ec4fc19d0d45274187488031ad1fc29758","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-04-06 10:31:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm_test pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/3563/ : SUCCESS in 2m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e807d8a9324acbfa91e6810feb0a5c275727bd1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-04-06 10:32:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/1976/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4a5b9d09401e7b243b05c747669bdf29365f5e7a","author":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"date":"2017-04-06 10:36:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6bc35c1b347ea4e7c63a201e6cecc9dacbe44e54","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2017-04-06 11:31:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-f47a8cc346186d894df7ffc1bb09b9d849816069 : FAILURE in 1h 03m 12s","accounts_in_message":[],"_revision_number":1},{"id":"48586794c49e3f248b5d773a04d9cba0839b3b9c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2017-04-06 11:47:26.000000000","message":"Abandoned\n\nduplicate of https://review.openstack.org/#/c/454037","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f47a8cc346186d894df7ffc1bb09b9d849816069","revisions":{"f47a8cc346186d894df7ffc1bb09b9d849816069":{"kind":"REWORK","_number":1,"created":"2017-04-06 10:28:25.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/89/454089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/89/454089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/89/454089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/89/454089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/89/454089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/89/454089/1"}}},"commit":{"parents":[{"commit":"33437668c35ebdbaa0b7e13e230ae3cfeb027705","subject":"Merge \"Update oslo.messaging to 5.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/33437668c35ebdbaa0b7e13e230ae3cfeb027705"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2017-04-06 10:27:53.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2017-04-06 10:27:53.000000000","tz":120},"subject":"Bump openstack-macros version","message":"Bump openstack-macros version\n\nCommit 4e5e178a5598 forgot to bump the version.\n\nChange-Id: I001446da2dfb09098d55cc9bdad5b003ba01a0c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/f47a8cc346186d894df7ffc1bb09b9d849816069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/f47a8cc346186d894df7ffc1bb09b9d849816069"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"33437668c35ebdbaa0b7e13e230ae3cfeb027705","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
