)]}'
{"id":"openstack%2Frpm-packaging~459448","triplet_id":"openstack%2Frpm-packaging~master~I64aa8ba1abd64af80c25c3d0ddc01d2a0fc7b1e1","project":"openstack/rpm-packaging","branch":"master","topic":"monasca_api","hashtags":[],"change_id":"I64aa8ba1abd64af80c25c3d0ddc01d2a0fc7b1e1","subject":"Update monasca-api to 2.0.0","status":"MERGED","created":"2017-04-24 20:39:15.000000000","updated":"2017-04-27 11:35:06.000000000","submitted":"2017-04-27 11:33:09.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d3b2347262494169708a5c5a18c9c6956f9a7e2","_number":459448,"virtual_id_number":459448,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2017-04-24 21:15:51.000000000","_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-04-24 21:16:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2017-04-27 11:33:09.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"all":[{"value":0,"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-04-25 07:40:34.000000000","_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2017-04-27 11:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2017-04-27 11:31:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-04-24 20:41:39.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-04-24 21:15:51.000000000","updated_by":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-04-24 21:16:38.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-04-25 07:40:34.000000000","updated_by":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"reviewer":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"state":"REVIEWER"},{"updated":"2017-04-27 11:31:29.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2017-04-27 11:33:09.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"4aa13d7f7741901d78421409b9160fca14eb99a8","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2017-04-24 20:39:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88c1ca68fe693ade94e0994e99ce186262431733","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-04-24 20:39:21.000000000","message":"Patch Set 1:\n\nBuild started.\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":1},{"id":"a128a2afaffa19a3f95dddd3e89c0e0b87ab5d90","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-04-24 20:40:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm_test pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/3776/ : FAILURE in 1m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4bce2395a84f1e6f815fbbfa612424d276e68287","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-24 20:41:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/48/459448/1/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/bbcac73/ : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1},{"id":"1294e02c16394eaad00687f2f45d791b6077f4b9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-04-24 20:41:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/2104/ : FAILURE in 2m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7c89fff755ca30cba6f4f572b97d927c55ac5096","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2017-04-24 20:43:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-414f7abb8e881d9d332abdfda8d01f8f05d11e1e : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6214d7d4fa95015a1f2a5b9208a1db42e9485f19","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2017-04-24 21:12:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa955dc24e8b55968f96aeb1681987be57b8ca84","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-04-24 21:12:10.000000000","message":"Patch Set 2:\n\nBuild started.\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":2},{"id":"2151406227dc5fca8e4e1e9fe6ed1645413abadf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-24 21:13:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/48/459448/2/check/gate-rpm-packaging-tox-lint-ubuntu-xenial/e21dd54/ : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2},{"id":"d8428e1e2eea7e53bbfc9dda68fb74302aea1029","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-04-24 21:14:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/2105/ : FAILURE in 2m 13s","accounts_in_message":[],"_revision_number":2},{"id":"fb67a2f7049e6037b6ca9728c1a84abff38bb68f","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-04-24 21:15:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (rpm_test pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/3777/ : SUCCESS in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5930b2927cdb9ed3c25fd81e31c6283f96b57522","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2017-04-24 21:16:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-2434c28f27f9194acc8eefd5438856798eca37b6 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d41deb8f89e46974a20b57f6638b10605142dcaf","author":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"date":"2017-04-25 07:40:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b9a90366f17e4b857895e2b14f99e19e36761faf","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2017-04-25 14:07:26.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"606b5b1b44b081af6aa7759ccd1b6d648e9791fa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-04-25 14:13:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/2118/ : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4c193dc36d204d2c4c17b4b904979c0bffb289e8","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2017-04-27 11:31:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"19e0879e2520d598f1a2b6a5e4a38470b5b00107","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-27 11:31:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"a8f9426bd7c6ef233ae63401fb7c99153f03065b","date":"2017-04-27 11:33:09.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"14a88a5d4ed9b71516080aaad83a3f05cf0b5fd3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-27 11:33:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-rpm-packaging-tox-lint-ubuntu-xenial http://logs.openstack.org/48/459448/2/gate/gate-rpm-packaging-tox-lint-ubuntu-xenial/5cc9090/ : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7008257effe33067b85b47e35dd684e6e69ed1e6","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-04-27 11:33:16.000000000","message":"Patch Set 2:\n\nBuild started.\nhttp://gate.fuel-infra.org\n\n\n- master-rpm-packaging-build-centos7 http://gate.fuel-infra.org : QUEUED","accounts_in_message":[],"_revision_number":2},{"id":"66ce475502592873782e11b837cd149697b2cbd7","author":{"_account_id":20656,"name":"Fuel Packaging CI","email":"ci+fuel-packaging@mirantis.com","username":"fuel-packaging-ci","tags":["SERVICE_USER"]},"date":"2017-04-27 11:35:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (rpm_publish pipeline).\n\n- master-rpm-packaging-build-centos7 https://packaging-ci.fuel-infra.org/job/master-rpm-packaging-build-centos7/3813/ : SUCCESS in 1m 47s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2434c28f27f9194acc8eefd5438856798eca37b6","revisions":{"414f7abb8e881d9d332abdfda8d01f8f05d11e1e":{"kind":"REWORK","_number":1,"created":"2017-04-24 20:39:15.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/48/459448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/48/459448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/48/459448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/48/459448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/48/459448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/48/459448/1"}}},"commit":{"parents":[{"commit":"26fc9226df06c57f937732b61b3325e887b2d2f5","subject":"Merge \"Update keystone to 12.0.0.0b1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/26fc9226df06c57f937732b61b3325e887b2d2f5"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2017-04-24 20:37:35.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2017-04-24 20:39:08.000000000","tz":120},"subject":"Update monasca-api to 2.0.0","message":"Update monasca-api to 2.0.0\n\nChange-Id: I64aa8ba1abd64af80c25c3d0ddc01d2a0fc7b1e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/414f7abb8e881d9d332abdfda8d01f8f05d11e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/414f7abb8e881d9d332abdfda8d01f8f05d11e1e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"26fc9226df06c57f937732b61b3325e887b2d2f5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2434c28f27f9194acc8eefd5438856798eca37b6":{"kind":"REWORK","_number":2,"created":"2017-04-24 21:12:04.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/48/459448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/48/459448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/48/459448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/48/459448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/48/459448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/48/459448/2"}}},"commit":{"parents":[{"commit":"26fc9226df06c57f937732b61b3325e887b2d2f5","subject":"Merge \"Update keystone to 12.0.0.0b1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/26fc9226df06c57f937732b61b3325e887b2d2f5"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2017-04-24 20:37:35.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2017-04-24 21:11:55.000000000","tz":120},"subject":"Update monasca-api to 2.0.0","message":"Update monasca-api to 2.0.0\n\nChange-Id: I64aa8ba1abd64af80c25c3d0ddc01d2a0fc7b1e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/2434c28f27f9194acc8eefd5438856798eca37b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/2434c28f27f9194acc8eefd5438856798eca37b6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"26fc9226df06c57f937732b61b3325e887b2d2f5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
