)]}'
{"id":"openstack%2Frpm-packaging~519954","triplet_id":"openstack%2Frpm-packaging~master~I087a406b182f16d07dbae59b8df4e4fbdeae640b","project":"openstack/rpm-packaging","branch":"master","topic":"tripleo-common","hashtags":[],"change_id":"I087a406b182f16d07dbae59b8df4e4fbdeae640b","subject":"Update tripleo-common to 7.6.4","status":"ABANDONED","created":"2017-11-15 03:07:47.000000000","updated":"2017-11-22 05:14:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d08c68d20cb085de53157d3ad3ed65dac34d23e7","_number":519954,"virtual_id_number":519954,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2017-11-16 15:30:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":-1,"date":"2017-11-16 13:57:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2017-11-16 13:58:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":-2,"date":"2017-11-21 09:10:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-11-16 13:36:12.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2017-11-16 13:57:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-16 13:58:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-16 15:30:31.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-21 09:10:22.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"}],"messages":[{"id":"e08a7acf139af37a2f637bf4110525d322ec07d1","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2017-11-15 03:07:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53f7a08fbebae04ef82e26e0a8fc56476786ca83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-15 03:11:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/54/519954/1/check/openstack-tox-linters/02d4921/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3cac266d132236af66fe94ac05e3e15e5515d0cd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-11-15 03:12:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/4477/ : FAILURE in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bde7911a9ce52424a509858fa6ea4c5d5010e5a5","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2017-11-15 03:23:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-f2dfe0d00bf4add29bf309b62620096cdc441191 : FAILURE in 15m 32s","accounts_in_message":[],"_revision_number":1},{"id":"efab92a9606b50b632839e7f2694558bd6c6a4ff","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2017-11-16 13:36:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"22bb81fc2e856e7222cbcfca88084f1559343bac","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-11-16 13:43:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/4498/ : FAILURE in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b0281975c42821819ddba68e1ce31d4d2f6c2f28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-16 13:46:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/54/519954/2/check/openstack-tox-linters/861f832/ : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":2},{"id":"91d2aa4e679121af8459146c1ec78a47078db2c5","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2017-11-16 13:51:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e4100afce5d569653260a70ffe571cca8570fa03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-16 13:57:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters http://logs.openstack.org/54/519954/3/check/openstack-tox-linters/3f1aad6/ : FAILURE in 4m 09s","accounts_in_message":[],"_revision_number":3},{"id":"6f46c7bfc52dfa7479cbc9c0b47c6f67b5df5d70","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-11-16 13:58:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- DLRN-rpmbuild-rpm-packaging https://review.rdoproject.org/jenkins/job/DLRN-rpmbuild-rpm-packaging/4499/ : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a4b4a06296328918b01388e132bca06574bdd557","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2017-11-16 15:30:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-d41e401399732a52d69d493748b3c34beab51275 : FAILURE in 1h 38m 54s","accounts_in_message":[],"_revision_number":3},{"id":"a911be5d14a3da079a7ff5e06ab45a8583b4004c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2017-11-21 09:10:22.000000000","message":"Patch Set 3: Code-Review-2\n\nthis is a change for stable/pike, not for master","accounts_in_message":[],"_revision_number":3},{"id":"1921d23dc18198fd6a28fbd1b0bc147d1516c4ad","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2017-11-22 05:14:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d41e401399732a52d69d493748b3c34beab51275","revisions":{"f2dfe0d00bf4add29bf309b62620096cdc441191":{"kind":"REWORK","_number":1,"created":"2017-11-15 03:07:47.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/54/519954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/54/519954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/1"}}},"commit":{"parents":[{"commit":"9624c9f26172251288b66c179fb9760bdf269427","subject":"Merge \"Update mistral-lib to 0.3.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9624c9f26172251288b66c179fb9760bdf269427"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2017-11-15 03:07:22.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2017-11-15 03:07:22.000000000","tz":480},"subject":"Update tripleo-common to 7.6.4","message":"Update tripleo-common to 7.6.4\n\nChange-Id: I087a406b182f16d07dbae59b8df4e4fbdeae640b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/f2dfe0d00bf4add29bf309b62620096cdc441191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/f2dfe0d00bf4add29bf309b62620096cdc441191"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9624c9f26172251288b66c179fb9760bdf269427","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7956beb7218203b3c65567920037a18acdcefcfd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2017-11-16 13:36:12.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/54/519954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/54/519954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/2"}}},"commit":{"parents":[{"commit":"9624c9f26172251288b66c179fb9760bdf269427","subject":"Merge \"Update mistral-lib to 0.3.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9624c9f26172251288b66c179fb9760bdf269427"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2017-11-15 03:07:22.000000000","tz":480},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2017-11-16 13:35:01.000000000","tz":60},"subject":"Update tripleo-common to 7.6.4","message":"Update tripleo-common to 7.6.4\n\nThis includes creating a new subpackage,\nopenstack-tripleo-common-container-base, to contain the dependencies\nrequired for the base TripleO container image.\n\nChange-Id: I087a406b182f16d07dbae59b8df4e4fbdeae640b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7956beb7218203b3c65567920037a18acdcefcfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7956beb7218203b3c65567920037a18acdcefcfd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9624c9f26172251288b66c179fb9760bdf269427","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d41e401399732a52d69d493748b3c34beab51275":{"kind":"REWORK","_number":3,"created":"2017-11-16 13:51:19.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/54/519954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/54/519954/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/54/519954/3"}}},"commit":{"parents":[{"commit":"6c12df1a97b6d6e9e24bcd2f33c31014bc174f4d","subject":"Merge \"Update reno to 2.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6c12df1a97b6d6e9e24bcd2f33c31014bc174f4d"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2017-11-15 03:07:22.000000000","tz":480},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2017-11-16 13:51:07.000000000","tz":60},"subject":"Update tripleo-common to 7.6.4","message":"Update tripleo-common to 7.6.4\n\nThis includes creating a new subpackage,\nopenstack-tripleo-common-container-base, to contain the dependencies\nrequired for the base TripleO container image.\n\nChange-Id: I087a406b182f16d07dbae59b8df4e4fbdeae640b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d41e401399732a52d69d493748b3c34beab51275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d41e401399732a52d69d493748b3c34beab51275"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6c12df1a97b6d6e9e24bcd2f33c31014bc174f4d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
