)]}'
{"id":"openstack%2Frpm-packaging~585954","triplet_id":"openstack%2Frpm-packaging~master~If5929a1e0bd7bb6a9f53863c6befde26a9417d9d","project":"openstack/rpm-packaging","branch":"master","topic":"nova","hashtags":[],"change_id":"If5929a1e0bd7bb6a9f53863c6befde26a9417d9d","subject":"nova: update to sphinx-build macro","status":"MERGED","created":"2018-07-26 05:44:48.000000000","updated":"2019-08-05 18:10:18.000000000","submitted":"2019-08-05 18:10:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"585954-1565028618003-e5490da1","meta_rev_id":"4b0495631dcffb0b35876a80f81df33b813e93c6","_number":585954,"virtual_id_number":585954,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-02 21:51:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-08-05 18:10:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-02 20:58:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-08-05 17:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-05 18:04:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 20:58:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-02 21:51:33.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-05 17:21:51.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-08-05 18:10:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fabdc092e480df586bf49ec95d7302880fb89195","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2018-07-26 05:44:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aae046bb9bcea5b50c9da7ea963649049a85a5f4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-26 05:53:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/54/585954/1/openstack-check/legacy-DLRN-rpmbuild-rpm-packaging-centos/1800f57/ : FAILURE in 7m 15s\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/54/585954/1/openstack-check/legacy-DLRN-rpmbuild-rpm-packaging-fedora/a5237f0/ : FAILURE in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b037d88d5f4a659be6dd34bb5d3861d5eb6303b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-26 05:54:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/54/585954/1/check/openstack-tox-linters/c7116f8/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e5e491c5443ba11132c437967b446331900e5144","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2018-07-26 06:10:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-16277b7bc4a04bc33d2112b06a1d25e742d119ea : FAILURE in 24m 52s","accounts_in_message":[],"_revision_number":1},{"id":"772279d2a69d659e83223a7f8aac549d747605a9","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2018-11-07 15:26:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c6fb42b37afab65ab519e272480c9c4073b14cf","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-03-05 09:36:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2800e5a4d4eaccf5ab30ee6e3044cd88495cd06e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 09:41:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/54/585954/2/check/openstack-tox-linters/95182ba/ : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"3a1a75f2f90f7f8128abb21b5479592d50d8540f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-05 09:48:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/54/585954/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/dbec0e4/ : FAILURE in 11m 15s\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/54/585954/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/72d02aa/ : FAILURE in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5da500f6976a7a8bae1de214f9ab66d9949a4bde","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-03-05 10:12:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-d29d7b5c450e5d5b7acb33599843fdf18f1573f8 : FAILURE in 13m 30s","accounts_in_message":[],"_revision_number":2},{"id":"a08d013ace8351eb2f2bee30a81c85a2abd420bc","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-04-12 18:03:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7f90d45097d1ccf007317d72211acd838bf16b93","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-12 18:11:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/54/585954/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/f93c582/ : FAILURE in 6m 50s\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/54/585954/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/aa38ab8/ : FAILURE in 6m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2cfe3060315a9083451799c041719744affc7db2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 18:12:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/54/585954/3/check/openstack-tox-linters/4afcd2c/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"c6652f5f42eca0ffa000462fd529ee7a9f1683b8","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 18:26:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-b1021fcaf82f390712db9028992b83dc012eaaff : FAILURE in 22m 43s","accounts_in_message":[],"_revision_number":3},{"id":"565f17b0634ea68437b7b8c9fdb51ceaedd3dee2","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-02 20:34:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"11b98fb164e6c15122a7cd1047da2ef7e3de7301","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 20:42:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/54/585954/4/check/openstack-tox-linters/eb6fb68/ : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":4},{"id":"c092450c6599a5f4141cf9424704bd49f9337acc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 20:58:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/54/585954/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/150330d/ : SUCCESS in 22m 48s","accounts_in_message":[],"_revision_number":4},{"id":"0383c8db953ba33aba09c2bbddc9f9a20bce3d2a","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 21:51:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-2648fbaf1f1cd4110ce4141c628ab411f6500541 : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":4},{"id":"1f296827d295922f75049b0fe7f7c873b45b1762","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-05 17:21:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3b7fc337b95aed7fb9b675893761afb2d919f1a7","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-05 18:04:41.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c94a3f04571a0e3c86beb5014710baf42e6d2cc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 18:04:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b61969b381cc00c6232a0ca6a14f7c6bf64ab3bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 18:10:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/54/585954/4/gate/openstack-tox-linters/0ec9718/ : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":4},{"id":"9cb9b4988fe4aea9d690941db4fcb145d4abb950","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 18:10:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2648fbaf1f1cd4110ce4141c628ab411f6500541","revisions":{"16277b7bc4a04bc33d2112b06a1d25e742d119ea":{"kind":"REWORK","_number":1,"created":"2018-07-26 05:44:48.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/54/585954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/54/585954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/1"}}},"commit":{"parents":[{"commit":"6f92b0ab69f6c5ff1eedd82604b54f3a3602c8d5","subject":"Update pypi url to new url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6f92b0ab69f6c5ff1eedd82604b54f3a3602c8d5"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2018-07-26 05:44:36.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2018-07-26 05:44:36.000000000","tz":120},"subject":"nova: update to sphinx-build","message":"nova: update to sphinx-build\n\nChange-Id: If5929a1e0bd7bb6a9f53863c6befde26a9417d9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/16277b7bc4a04bc33d2112b06a1d25e742d119ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/16277b7bc4a04bc33d2112b06a1d25e742d119ea"}]},"branch":"refs/heads/master"},"d29d7b5c450e5d5b7acb33599843fdf18f1573f8":{"kind":"REWORK","_number":2,"created":"2019-03-05 09:36:30.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/54/585954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/54/585954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/2"}}},"commit":{"parents":[{"commit":"f55faac5557d5f6b6cc6ce3759134b988b149161","subject":"Merge \"oslo.messaging: Update to 9.5.0 and Remove zeromq dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/f55faac5557d5f6b6cc6ce3759134b988b149161"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2018-07-26 05:44:36.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-03-05 09:36:22.000000000","tz":60},"subject":"nova: update to sphinx-build","message":"nova: update to sphinx-build\n\nChange-Id: If5929a1e0bd7bb6a9f53863c6befde26a9417d9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d29d7b5c450e5d5b7acb33599843fdf18f1573f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d29d7b5c450e5d5b7acb33599843fdf18f1573f8"}]},"branch":"refs/heads/master"},"b1021fcaf82f390712db9028992b83dc012eaaff":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-12 18:03:31.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/54/585954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/54/585954/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/3"}}},"commit":{"parents":[{"commit":"d1d8172e0b00c11acb6767203926bdab6fe83a9a","subject":"Merge \"cinder: Cleanup spec template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d1d8172e0b00c11acb6767203926bdab6fe83a9a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2018-07-26 05:44:36.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-04-12 18:03:31.000000000","tz":0},"subject":"nova: update to sphinx-build","message":"nova: update to sphinx-build\n\nChange-Id: If5929a1e0bd7bb6a9f53863c6befde26a9417d9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b1021fcaf82f390712db9028992b83dc012eaaff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b1021fcaf82f390712db9028992b83dc012eaaff"}]},"branch":"refs/heads/master"},"2648fbaf1f1cd4110ce4141c628ab411f6500541":{"kind":"REWORK","_number":4,"created":"2019-08-02 20:34:42.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/54/585954/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/54/585954/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/54/585954/4"}}},"commit":{"parents":[{"commit":"b9a6ec6f782cfb8bd67e4a56d53a949b6b4262ee","subject":"Merge \"nova: Remove nova-consoleauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b9a6ec6f782cfb8bd67e4a56d53a949b6b4262ee"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2018-07-26 05:44:36.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-02 20:34:25.000000000","tz":120},"subject":"nova: update to sphinx-build macro","message":"nova: update to sphinx-build macro\n\nChange-Id: If5929a1e0bd7bb6a9f53863c6befde26a9417d9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/2648fbaf1f1cd4110ce4141c628ab411f6500541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/2648fbaf1f1cd4110ce4141c628ab411f6500541"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
