)]}'
{"id":"openstack%2Frpm-packaging~656431","triplet_id":"openstack%2Frpm-packaging~master~I19a50033f41dd9b043ae37afb1c47c667df12c21","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"I19a50033f41dd9b043ae37afb1c47c667df12c21","subject":"Update to opendev","status":"MERGED","created":"2019-04-30 02:54:03.000000000","updated":"2019-05-03 16:17:00.000000000","submitted":"2019-05-03 16:17:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656431-1556900220829-eab9e842","meta_rev_id":"d3763ee955c8c353d2277998f74d5d8670f59044","_number":656431,"virtual_id_number":656431,"owner":{"_account_id":29979,"name":"hutianhao27","email":"hutianhao27@126.com","username":"hth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-01 03:01:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":29979,"name":"hutianhao27","email":"hutianhao27@126.com","username":"hth"},{"value":2,"date":"2019-05-03 16:17:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-01 02:43:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-03 15:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":29979,"name":"hutianhao27","email":"hutianhao27@126.com","username":"hth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-03 16:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":29979,"name":"hutianhao27","email":"hutianhao27@126.com","username":"hth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-03 16:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29979,"name":"hutianhao27","email":"hutianhao27@126.com","username":"hth"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-01 02:43:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-01 03:01:38.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-03 15:37:25.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-03 16:07:45.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-05-03 16:17:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2773079a4de96ca2d8a62d3d5d75ce3e1030543","author":{"_account_id":29979,"name":"hutianhao27","email":"hutianhao27@126.com","username":"hth"},"date":"2019-04-30 02:54:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10ab44807d4fda2225350858d7ccec4175a116a1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-30 02:57:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/31/656431/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/ec0c20c/ : SUCCESS in 2m 27s\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/31/656431/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/978bf57/ : SUCCESS in 2m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09c808769bbd00c88cce2b5af720d76710fd8b90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 03:02:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/31/656431/1/check/openstack-tox-linters/d02e86d/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d1b627f77253b1d2e186bcb4c19e2c4f3e1dcfae","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 03:11:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-09e8870fdb46eaa0773e5605a9b59c94d58972f0 : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9340555be8e63ce66f4813260d26081f9b272e71","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-30 19:39:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"feead787b10829d68db330a964d4ccb615c874e0","author":{"_account_id":29979,"name":"hutianhao27","email":"hutianhao27@126.com","username":"hth"},"date":"2019-05-01 02:40:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a9c090ae5b320c9a996c723e6728ef72c8c2940c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-01 02:43:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/31/656431/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/de2943c/ : SUCCESS in 2m 03s\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/31/656431/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/6fefc08/ : SUCCESS in 2m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4643cdacd6ab3ebe61d4dc813c79a421fdedce21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 02:44:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/31/656431/2/check/openstack-tox-linters/cb9c4e7/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"0d33fca70ad053b4a9462f924424d3c2b90dba56","author":{"_account_id":29979,"name":"hutianhao27","email":"hutianhao27@126.com","username":"hth"},"date":"2019-05-01 02:45:08.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThank you for your review.","accounts_in_message":[],"_revision_number":1},{"id":"a9c405405ef262f4c0bfd6c762c80bdc8c6f80b6","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-05-01 03:01:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-121f4013cfc54aa20140b71e4351666fdb7eab50 : SUCCESS in 20m 43s","accounts_in_message":[],"_revision_number":2},{"id":"90ce7573330b9b7e5f9e0633abb11cc179bd3cfb","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-03 15:37:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f403b9ce2838dc7c6fd545993aedda329065259b","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-05-03 16:07:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8852f3aa2e9aa98bfb22e8d98011eff3239b8e89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:07:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e15e539bb3c23e3b82ad46f0e6c46c7a86d0e9c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:17:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/31/656431/2/gate/openstack-tox-linters/ea00da8/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d3763ee955c8c353d2277998f74d5d8670f59044","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:17:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b3651b9858d997d4f364db390d6a7ac182dba2cf","revisions":{"09e8870fdb46eaa0773e5605a9b59c94d58972f0":{"kind":"REWORK","_number":1,"created":"2019-04-30 02:54:03.000000000","uploader":{"_account_id":29979,"name":"hutianhao27","email":"hutianhao27@126.com","username":"hth"},"ref":"refs/changes/31/656431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/31/656431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/656431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/656431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/656431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/31/656431/1"}}},"commit":{"parents":[{"commit":"aac9a8ee8378aee97a6fef92dd1a07baf44bbf7b","subject":"update pyghmi to 1.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/aac9a8ee8378aee97a6fef92dd1a07baf44bbf7b"}]}],"author":{"name":"hutianhao","email":"hu.tianhao@99cloud.net","date":"2019-04-30 02:17:45.000000000","tz":0},"committer":{"name":"hutianhao","email":"hu.tianhao@99cloud.net","date":"2019-04-30 02:17:45.000000000","tz":0},"subject":"Update to opendev","message":"Update to opendev\n\nChange-Id: I19a50033f41dd9b043ae37afb1c47c667df12c21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/09e8870fdb46eaa0773e5605a9b59c94d58972f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/09e8870fdb46eaa0773e5605a9b59c94d58972f0"}]},"branch":"refs/heads/master"},"b3651b9858d997d4f364db390d6a7ac182dba2cf":{"kind":"REWORK","_number":2,"created":"2019-05-01 02:40:35.000000000","uploader":{"_account_id":29979,"name":"hutianhao27","email":"hutianhao27@126.com","username":"hth"},"ref":"refs/changes/31/656431/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/31/656431/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/656431/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/656431/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/656431/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/31/656431/2"}}},"commit":{"parents":[{"commit":"aac9a8ee8378aee97a6fef92dd1a07baf44bbf7b","subject":"update pyghmi to 1.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/aac9a8ee8378aee97a6fef92dd1a07baf44bbf7b"}]}],"author":{"name":"hutianhao","email":"hu.tianhao@99cloud.net","date":"2019-04-30 02:17:45.000000000","tz":0},"committer":{"name":"hutianhao","email":"hu.tianhao@99cloud.net","date":"2019-05-01 02:39:53.000000000","tz":0},"subject":"Update to opendev","message":"Update to opendev\n\nChange-Id: I19a50033f41dd9b043ae37afb1c47c667df12c21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b3651b9858d997d4f364db390d6a7ac182dba2cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b3651b9858d997d4f364db390d6a7ac182dba2cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
