)]}'
{"id":"openstack%2Frpm-packaging~656438","triplet_id":"openstack%2Frpm-packaging~master~Icce64c2ce1b517452e094bfd8ca7fd5096de4de2","project":"openstack/rpm-packaging","branch":"master","topic":"rpm-packaging","hashtags":[],"change_id":"Icce64c2ce1b517452e094bfd8ca7fd5096de4de2","subject":"Move to opendev","status":"ABANDONED","created":"2019-04-30 03:46:42.000000000","updated":"2019-04-30 05:42:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"61e60cfd2d01156ea0c2ceb98a72b792bff3d534","_number":656438,"virtual_id_number":656438,"owner":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-04-30 04:07:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-30 03:50:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-30 03:51:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-30 03:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 03:51:46.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 04:07:21.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17c97413fd608e0aa3075d07f6a38bcad197bab3","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-04-30 03:46:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74d9040e63f791eaef696fa9d2445bfa09811600","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 03:50:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/38/656438/1/check/openstack-tox-linters/7ea68bf/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"71acda2f680a8482d623671bc8af3db2da5aa970","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-30 03:51:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/38/656438/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/d3bba04/ : SUCCESS in 2m 19s\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/38/656438/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/945581a/ : SUCCESS in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1755bb959dcf1a033028464bb64130a8ada8481d","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 04:07:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-f5cd3f34ec3680e7b5904c120f6598fc5b4c3351 : SUCCESS in 20m 19s","accounts_in_message":[],"_revision_number":1},{"id":"21e6ceb7928a97cf9bc021ddd34a756a190bfa55","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-04-30 05:42:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b33486a956230e1c1a7272130bfa70fa2fe4be73","revisions":{"b33486a956230e1c1a7272130bfa70fa2fe4be73":{"kind":"REWORK","_number":1,"created":"2019-04-30 03:46:42.000000000","uploader":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"ref":"refs/changes/38/656438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/38/656438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/38/656438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/38/656438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/38/656438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/38/656438/1"}}},"commit":{"parents":[{"commit":"c8cfa3b19f935801f345143050bde92b984ef43d","subject":"Merge \"update python-tripleoclient to 11.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/c8cfa3b19f935801f345143050bde92b984ef43d"}]}],"author":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-04-30 02:17:32.000000000","tz":480},"committer":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-04-30 02:17:32.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: Icce64c2ce1b517452e094bfd8ca7fd5096de4de2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b33486a956230e1c1a7272130bfa70fa2fe4be73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b33486a956230e1c1a7272130bfa70fa2fe4be73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
