)]}'
{"id":"openstack%2Frpm-packaging~666622","triplet_id":"openstack%2Frpm-packaging~master~Ie821e88003b554ca97bdbc605e71633097a3c3b0","project":"openstack/rpm-packaging","branch":"master","topic":"py3-keystone","hashtags":[],"change_id":"Ie821e88003b554ca97bdbc605e71633097a3c3b0","subject":"Use python3 for doc build","status":"MERGED","created":"2019-06-20 16:00:33.000000000","updated":"2019-07-23 09:44:59.000000000","submitted":"2019-07-23 09:44:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666622-1563875099260-0ae53db0","meta_rev_id":"66beb426b23455f696a9ea9f1285aa25018536f9","_number":666622,"virtual_id_number":666622,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-07-22 22:26:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-07-23 09:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-22 22:09:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-23 06:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-23 09:36:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-21 08:16:38.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-07-03 07:05:40.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-07-22 22:09:23.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-22 22:26:52.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-23 09:36:56.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-07-23 09:44:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b8d676ab7aa91750a1e24e8f9ac54106ac029ca","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-20 16:00:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"573ea39d4b78c7396319ecb5773e7bc1387958ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-20 16:08:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/c038b34/ : FAILURE in 6m 48s","accounts_in_message":[],"_revision_number":1},{"id":"257fbb449a00973b76f05b424e7fd90604daeb43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 16:16:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/1/check/openstack-tox-linters/912db74/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e218c2a80611abba4680f9114dc67d07538db97f","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-06-20 16:18:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e5fa6c274b4a2a554e18dfd32250d3ceedafd73","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-20 16:32:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-0563871937491d0607be6872cf9c0af9678ebe99 : FAILURE in 31m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5254eca74126eb75b60ac00231ed1e1705928bff","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-20 16:40:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a3611cc521243da0aed7ddba76775b69e5d65249","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-20 16:47:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/4edfba9/ : FAILURE in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"a8e3cdb677b187d48df4525d4b4d8b38dceb51ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 16:53:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/2/check/openstack-tox-linters/8c2e0a0/ : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":2},{"id":"addfbb8b6da6f29dd6b2bb470a8eb90c17e7699b","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-20 17:22:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-89bd49dcffe58f075683ae29535bb7df79ed2283 : FAILURE in 41m 44s","accounts_in_message":[],"_revision_number":2},{"id":"7f24373afd0f707078df9ee4fff0fc676bab6720","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-06-21 08:16:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35640756d4fc4dba7f22ffed2c4ca11cc63d5ebc","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-21 16:45:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"856645ba585766dcea7e45618cb6228caedf1a7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 16:54:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/3/check/openstack-tox-linters/ea7f6e9/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":3},{"id":"ba286c1ea3a9e3e7be1121f4aa1647a5da53205a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-21 17:20:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/ef2026f/ : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":3},{"id":"f2f8cbf9398d8be52752e273278427b7c57108d8","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-21 18:09:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-779319b1344bc0de6c67e9553753a5deddd6f126 : FAILURE in 1h 24m 29s","accounts_in_message":[],"_revision_number":3},{"id":"5d1417cf02992878dce95bf8395c1bf0713c42f0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-21 19:35:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ef9036589685eff46663fc9d62b9315676f39ae8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 19:45:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/4/check/openstack-tox-linters/0773f19/ : FAILURE in 3m 38s","accounts_in_message":[],"_revision_number":4},{"id":"b3c5282e23809a653b659b016b7350bc5ab27534","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-21 20:14:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/c486cbc/ : FAILURE in 37m 28s","accounts_in_message":[],"_revision_number":4},{"id":"adb69b817efb7ef7083f16cbe28cdb013d89d2c0","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-21 20:16:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-89c41862992e820df2fe75d2a0d5f3b0e4f06b9b : FAILURE in 40m 22s","accounts_in_message":[],"_revision_number":4},{"id":"a935536dac3c7bf6ea2096b6adc9f71b6a736a9a","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-21 21:07:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"103036fc7573824a58222a032af492b9ca98cdbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 21:18:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/5/check/openstack-tox-linters/b7b03d1/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":5},{"id":"abc6d2fab6a1f82d98cd94c410b9214cd9553c37","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-21 21:37:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/5/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/ea56697/ : SUCCESS in 28m 22s","accounts_in_message":[],"_revision_number":5},{"id":"e0988062f39a71e513211f806a386dcc7b712f5d","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-21 21:48:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-b092215e8c4201f9f33cb2dfff7c075e8129052e : FAILURE in 40m 01s","accounts_in_message":[],"_revision_number":5},{"id":"33e79b7b44655591dbb2ee901cd30afbf999dbc7","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-24 08:27:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4109bf9b00bce84722f5d2be3bc5cefc324d1223","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 08:36:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/6/check/openstack-tox-linters/274307c/ : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":6},{"id":"25fce943708632f01e4ab9d799e5fe942af7017d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-24 08:40:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/6/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/202ca44/ : FAILURE in 11m 47s","accounts_in_message":[],"_revision_number":6},{"id":"596b8ed8ed28acf80467fe4132f72ed91dc32c94","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-24 10:59:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-49012ace96c6e67fe71385a4f09664cd8f9404ca : FAILURE in 2h 31m 23s","accounts_in_message":[],"_revision_number":6},{"id":"d9442afafb8bb77d67bf4d07a3e51e3077f93a0c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-25 10:52:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4ab81c09b6a4c5f607db41a68a2e597c686b28ff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 10:56:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/7/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/be6925c/ : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":7},{"id":"0922c9e2634d451693c4dadd85b76ee9a4bfce53","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-25 10:59:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-ba055dec4c627a1e47f833604e1bb4952a4434d5 : FAILURE in 6m 24s","accounts_in_message":[],"_revision_number":7},{"id":"b432ccdd9684d423602a60037a84d3d2efb9401e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 11:23:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/7/check/openstack-tox-linters/01ef035/ : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":7},{"id":"2cbe8075a2c581d6f1dbec286875d1cf87dbbcfb","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-26 10:26:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"15f3f25c5090356fae3d117253d9c3497f099264","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 10:31:21.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/8/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/0bee6b6/ : FAILURE in 3m 27s","accounts_in_message":[],"_revision_number":8},{"id":"ca5390be0c932671dc37b13e935c819146ce0ba8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 10:31:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/8/check/openstack-tox-linters/5a3aba2/ : FAILURE in 3m 12s","accounts_in_message":[],"_revision_number":8},{"id":"1370cd87e3bc77a491c50e1f94bb4ec64b3aea4e","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-26 10:33:30.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-822f8a96e320e0453ba2c6b4d41917011f96073b : FAILURE in 6m 20s","accounts_in_message":[],"_revision_number":8},{"id":"fb9739bf4aa79713a389f086d66b0fae80539433","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-27 21:02:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"69b1c85a5b1c3e7fb873f6b95de1b8f1b1da2f28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 21:11:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/9/check/openstack-tox-linters/61a85b4/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":9},{"id":"b7cc600578cedaee1232ef5ef5b8e64507c591e7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 21:24:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/9/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/09c8af3/ : FAILURE in 21m 50s","accounts_in_message":[],"_revision_number":9},{"id":"aa9bc09b0e7c2d8c63b39bdc691b7b0db3ac5d6f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-27 21:32:22.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4858fe6a9172a27df8335a4a406cba6cb6a591f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 21:42:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/10/check/openstack-tox-linters/ad1f2f2/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":10},{"id":"973eede86c5959ab75f68eaf2ee5a8057e1b389e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 21:53:01.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/10/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/23abdb7/ : FAILURE in 19m 12s","accounts_in_message":[],"_revision_number":10},{"id":"cc59cf038e36b0c4f14b34c8526aec63af537a38","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-27 23:51:33.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-5993de13837c0f9ede761e997871c0fff530aa80 : FAILURE in 2h 18m 50s","accounts_in_message":[],"_revision_number":10},{"id":"0e0d5ecb05c26345ecb7faa2f5151ce5bdb3f0e9","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-28 13:34:53.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0002c330b3b8d068ed172c78386a7d39f0a3cf2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 13:41:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/11/check/openstack-tox-linters/fdfe065/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":11},{"id":"e9cc559cdfb6c9e9255a773cd106eb332e38618e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 13:54:24.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/11/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/c7c9f0d/ : FAILURE in 18m 50s","accounts_in_message":[],"_revision_number":11},{"id":"88ffad8dd2bd044a77442caaab36eb433bff967e","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-28 15:47:55.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-44cf58043ffc46c4589150a1fd10464ef032ce31 : FAILURE in 2h 12m 43s","accounts_in_message":[],"_revision_number":11},{"id":"1c4055159a74920638e6484a37e824de5deca207","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-28 17:20:59.000000000","message":"Patch Set 11: Workflow-1\n\nsorry for taking this over.. I\u0027ve splitted my stuff into additional reviews, hopefully we can get that one into a merging state and then rebase this on top.","accounts_in_message":[],"_revision_number":11},{"id":"d6e36cc8b519ceec995acc186f7afdff8f892e62","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-01 08:43:52.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"51e05df342c327c27221ef94cf715dc8fce31534","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 08:51:05.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/12/check/openstack-tox-linters/c3332d2/ : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":12},{"id":"ef460de5663420077a9df416c83bef423098a46e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 08:51:37.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/12/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/0c3b695/ : FAILURE in 7m 05s","accounts_in_message":[],"_revision_number":12},{"id":"df744a8cacdcc4ff30c9ff2b15f3906862d825b6","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-01 11:56:28.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-828e0e153e628af93a1f0490c314a658f992fe87 : SUCCESS in 3h 12m 17s","accounts_in_message":[],"_revision_number":12},{"id":"076fe3e919ab154f910b86b442f8f6c463168a52","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-01 16:11:43.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"dae1bcb844d61151fea70fde0e59bcb35c896fc3","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-01 16:11:49.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"a0866b7acd8c484b48cf4156cecd7ae95e77d1d1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 16:20:35.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/12/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/dc57b92/ : FAILURE in 8m 02s","accounts_in_message":[],"_revision_number":12},{"id":"7007befaed95cae5bead8b6abbd82decf97d8277","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 16:20:55.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/12/check/openstack-tox-linters/eb7bc38/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":12},{"id":"2678d1b68e1ae2180c952c9147fcd68120fa3910","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-01 16:58:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-1ffef192402406294136a2dc79009fb3ac9007ab : FAILURE in 46m 35s","accounts_in_message":[],"_revision_number":12},{"id":"ce21134a6d5fe4600b91c5a2dbc0bdc43348bc9d","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-02 04:01:24.000000000","message":"Patch Set 12: Code-Review-1\n\nstill failing in the SUSE CI with the error mentioned in the commit msg","accounts_in_message":[],"_revision_number":12},{"id":"654c6d24841a024bbc25f3807cae5e7e196dec44","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-02 20:33:17.000000000","message":"Patch Set 12: -Code-Review\n\nTom, are you sure? the error failing is with *nova*. this one fixes keystone. \n\nThe nova review is also under review, but its unrelated to this one. \n\nBTW, I think this was just a bug in the SUSE CI, which I just fixed. I don\u0027t think this is needed anymore.","accounts_in_message":[],"_revision_number":12},{"id":"a248b64067be3bdec2ccc0877365d7e92eb3096a","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-03 07:05:40.000000000","message":"Patch Set 12: -Code-Review\n\n\u003e Tom, are you sure? the error failing is with *nova*. this one fixes\n \u003e keystone.\n\nHm. Maybe I looked at an old log. Not sure.\n\n \u003e The nova review is also under review, but its unrelated to this\n \u003e one.\n \u003e \n \u003e BTW, I think this was just a bug in the SUSE CI, which I just\n \u003e fixed. I don\u0027t think this is needed anymore.\n\nSo should we abandon this one?","accounts_in_message":[],"_revision_number":12},{"id":"384d54ecc26e0c7e2955cd52f504e0e14f88284e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-22 21:26:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"6c02994028e4421464fbfd78ce64cc58234b6adf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 21:37:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/13/check/openstack-tox-linters/c9399d3/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":13},{"id":"663a05206c7aad17b2536971baf2bd9740c1189f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 22:09:23.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/22/666622/13/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/d5c2835/ : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":13},{"id":"1ffa5c4f80d1613f8fdf908462e007a83c58c3e1","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-22 22:26:52.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-82af7ff4ac6487c0f390cd12f1069f720182e2bf : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":13},{"id":"e7ab7c1822c85de52a9976fd455bb5604bce69fb","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-23 06:34:48.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6e6f866ef48fdc4368feece7d0db1ab60e110a1e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-23 09:36:56.000000000","message":"Patch Set 13: Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"94f1f04bfd0ec5ae1b3efffba7e345095116faf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 09:37:08.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"ce8f5f7c0a0d8c3c7a4114b9ce47dfe7305c5667","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 09:44:59.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/22/666622/13/gate/openstack-tox-linters/a9acb60/ : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":13},{"id":"dbf9e0f2d6e762152a71443a37e5abde44de3173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 09:44:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"82af7ff4ac6487c0f390cd12f1069f720182e2bf","revisions":{"0563871937491d0607be6872cf9c0af9678ebe99":{"kind":"REWORK","_number":1,"created":"2019-06-20 16:00:33.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/22/666622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/1"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"subject":"Use python3 for keystone doc builds","message":"Use python3 for keystone doc builds\n\nWithout this patch, the doc build for keystone fails with a Sphinx\nincompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0563871937491d0607be6872cf9c0af9678ebe99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0563871937491d0607be6872cf9c0af9678ebe99"}]},"branch":"refs/heads/master"},"89bd49dcffe58f075683ae29535bb7df79ed2283":{"kind":"REWORK","_number":2,"created":"2019-06-20 16:40:09.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/22/666622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/2"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 16:40:02.000000000","tz":-420},"subject":"Use python3 for keystone doc builds","message":"Use python3 for keystone doc builds\n\nWithout this patch, the doc build for keystone fails with a Sphinx\nincompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/89bd49dcffe58f075683ae29535bb7df79ed2283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/89bd49dcffe58f075683ae29535bb7df79ed2283"}]},"branch":"refs/heads/master"},"779319b1344bc0de6c67e9553753a5deddd6f126":{"kind":"REWORK","_number":3,"created":"2019-06-21 16:45:11.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/22/666622/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/3"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-21 16:44:59.000000000","tz":-420},"subject":"Use python3 for keystone doc builds","message":"Use python3 for keystone doc builds\n\nWithout this patch, the doc build for keystone fails with a Sphinx\nincompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/779319b1344bc0de6c67e9553753a5deddd6f126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/779319b1344bc0de6c67e9553753a5deddd6f126"}]},"branch":"refs/heads/master"},"89c41862992e820df2fe75d2a0d5f3b0e4f06b9b":{"kind":"REWORK","_number":4,"created":"2019-06-21 19:35:56.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/22/666622/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/4"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-21 19:34:42.000000000","tz":-420},"subject":"Use python3 for keystone, cinder, watcher docs","message":"Use python3 for keystone, cinder, watcher docs\n\nWithout this patch, the doc build for these packages fails with a Sphinx\nincompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nor:\n\n  Unknown event name: autodoc-process-docstring\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nFor cinder and watcher, we need to duplicate some of the build\nrequirements under the doc package now that they are python3 only.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/89c41862992e820df2fe75d2a0d5f3b0e4f06b9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/89c41862992e820df2fe75d2a0d5f3b0e4f06b9b"}]},"branch":"refs/heads/master"},"b092215e8c4201f9f33cb2dfff7c075e8129052e":{"kind":"REWORK","_number":5,"created":"2019-06-21 21:07:45.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/22/666622/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/5"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-21 21:05:37.000000000","tz":-420},"subject":"Use python3 for keystone doc build","message":"Use python3 for keystone doc build\n\nWithout this patch, the doc build for the keystone package fails with a\nSphinx incompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nFor cinder and watcher, we need to duplicate some of the build\nrequirements under the doc package now that they are python3 only.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b092215e8c4201f9f33cb2dfff7c075e8129052e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b092215e8c4201f9f33cb2dfff7c075e8129052e"}]},"branch":"refs/heads/master"},"49012ace96c6e67fe71385a4f09664cd8f9404ca":{"kind":"REWORK","_number":6,"created":"2019-06-24 08:27:42.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/666622/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/6"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-24 08:27:33.000000000","tz":120},"subject":"Use python3 for doc build","message":"Use python3 for doc build\n\nWithout this patch, the doc build for the keystone package fails with a\nSphinx incompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nFor cinder and watcher, we need to duplicate some of the build\nrequirements under the doc package now that they are python3 only.\n\nNOTE(dmllr): Add sphinx_build macro, fix other package builds along\nthe way.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/49012ace96c6e67fe71385a4f09664cd8f9404ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/49012ace96c6e67fe71385a4f09664cd8f9404ca"}]},"branch":"refs/heads/master"},"ba055dec4c627a1e47f833604e1bb4952a4434d5":{"kind":"REWORK","_number":7,"created":"2019-06-25 10:52:33.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/666622/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/7"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-25 10:52:24.000000000","tz":120},"subject":"Use python3 for doc build","message":"Use python3 for doc build\n\nWithout this patch, the doc build for the keystone package fails with a\nSphinx incompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nFor cinder and watcher, we need to duplicate some of the build\nrequirements under the doc package now that they are python3 only.\n\nNOTE(dmllr): Add sphinx_build macro, fix other package builds along\nthe way.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ba055dec4c627a1e47f833604e1bb4952a4434d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ba055dec4c627a1e47f833604e1bb4952a4434d5"}]},"branch":"refs/heads/master"},"822f8a96e320e0453ba2c6b4d41917011f96073b":{"kind":"REWORK","_number":8,"created":"2019-06-26 10:26:50.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/666622/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/8"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-26 10:26:41.000000000","tz":120},"subject":"Use python3 for doc build","message":"Use python3 for doc build\n\nWithout this patch, the doc build for the keystone package fails with a\nSphinx incompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nFor cinder and watcher, we need to duplicate some of the build\nrequirements under the doc package now that they are python3 only.\n\nNOTE(dmllr): Add sphinx_build macro, fix other package builds along\nthe way.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/822f8a96e320e0453ba2c6b4d41917011f96073b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/822f8a96e320e0453ba2c6b4d41917011f96073b"}]},"branch":"refs/heads/master"},"29d78ab527ab4ceb7fe95cf95c04451b0e22093a":{"kind":"REWORK","_number":9,"created":"2019-06-27 21:02:07.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/666622/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/9"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-27 21:01:56.000000000","tz":120},"subject":"Use python3 for doc build","message":"Use python3 for doc build\n\nWithout this patch, the doc build for the keystone package fails with a\nSphinx incompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nFor cinder and watcher, we need to duplicate some of the build\nrequirements under the doc package now that they are python3 only.\n\nNOTE(dmllr): Add sphinx_build macro, fix other package builds along\nthe way.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/29d78ab527ab4ceb7fe95cf95c04451b0e22093a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/29d78ab527ab4ceb7fe95cf95c04451b0e22093a"}]},"branch":"refs/heads/master"},"5993de13837c0f9ede761e997871c0fff530aa80":{"kind":"REWORK","_number":10,"created":"2019-06-27 21:32:22.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/666622/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/10"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-27 21:32:03.000000000","tz":120},"subject":"Use python3 for doc build","message":"Use python3 for doc build\n\nWithout this patch, the doc build for the keystone package fails with a\nSphinx incompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nFor cinder and watcher, we need to duplicate some of the build\nrequirements under the doc package now that they are python3 only.\n\nNOTE(dmllr): Add sphinx_build macro, fix other package builds along\nthe way.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5993de13837c0f9ede761e997871c0fff530aa80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5993de13837c0f9ede761e997871c0fff530aa80"}]},"branch":"refs/heads/master"},"44cf58043ffc46c4589150a1fd10464ef032ce31":{"kind":"REWORK","_number":11,"created":"2019-06-28 13:34:53.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/666622/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/11"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 13:33:39.000000000","tz":120},"subject":"Use python3 for doc build","message":"Use python3 for doc build\n\nWithout this patch, the doc build for the keystone package fails with a\nSphinx incompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nFor cinder and watcher, we need to duplicate some of the build\nrequirements under the doc package now that they are python3 only.\n\nNOTE(dmllr): Add sphinx_build macro, fix other package builds along\nthe way.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/44cf58043ffc46c4589150a1fd10464ef032ce31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/44cf58043ffc46c4589150a1fd10464ef032ce31"}]},"branch":"refs/heads/master"},"8d2916c7a36301d84bc15aaaec45728fa6f0f7c8":{"kind":"REWORK","_number":12,"created":"2019-07-01 08:43:52.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/666622/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/12"}}},"commit":{"parents":[{"commit":"6526a943b6a474984e9870104648b0e69f04af44","subject":"Merge \"renderspec: install renderspec binary only for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6526a943b6a474984e9870104648b0e69f04af44"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-01 08:43:33.000000000","tz":120},"subject":"Use python3 for doc build","message":"Use python3 for doc build\n\nWithout this patch, the doc build for the keystone package fails with a\nSphinx incompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8d2916c7a36301d84bc15aaaec45728fa6f0f7c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8d2916c7a36301d84bc15aaaec45728fa6f0f7c8"}]},"branch":"refs/heads/master"},"82af7ff4ac6487c0f390cd12f1069f720182e2bf":{"kind":"REWORK","_number":13,"created":"2019-07-22 21:26:27.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/22/666622/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/22/666622/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/22/666622/13"}}},"commit":{"parents":[{"commit":"259690d920b4fac1e7b64cd335d27d16c0e14539","subject":"Merge \"keystone: add new dependency for documentation build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/259690d920b4fac1e7b64cd335d27d16c0e14539"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-20 15:57:04.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-22 21:26:22.000000000","tz":-420},"subject":"Use python3 for doc build","message":"Use python3 for doc build\n\nWithout this patch, the doc build for the keystone package fails with a\nSphinx incompatibility:\n\n  AttributeError: \u0027Sphinx\u0027 object has no attribute \u0027add_css_file\u0027\n\nSphinx 2.x has dropped support for python2, so start using the python3\nversions of these packages which should get them all in sync.\n\nChange-Id: Ie821e88003b554ca97bdbc605e71633097a3c3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/82af7ff4ac6487c0f390cd12f1069f720182e2bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/82af7ff4ac6487c0f390cd12f1069f720182e2bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
