)]}'
{"id":"openstack%2Frpm-packaging~668209","triplet_id":"openstack%2Frpm-packaging~master~I7a9d8a8cbe508cee446fa1b7f512f510c2187c19","project":"openstack/rpm-packaging","branch":"master","topic":"get_things_green","hashtags":[],"change_id":"I7a9d8a8cbe508cee446fa1b7f512f510c2187c19","subject":"Switch to python3 for cinder","status":"MERGED","created":"2019-06-28 17:15:42.000000000","updated":"2019-08-21 07:27:52.000000000","submitted":"2019-08-21 07:27:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668209-1566372472188-462ff7de","meta_rev_id":"2d284a987c18c362fa96ff2d1e5ba44983ca13b9","_number":668209,"virtual_id_number":668209,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-13 20:00:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-08-21 07:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-13 19:59:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-21 07:16:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-21 07:16:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 15:41:35.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-08-13 19:59:33.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 20:00:19.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 07:16:48.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-08-21 07:27:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e76206e9348b7669486b5c109e740311f2106d8","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-28 17:15:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92bb7decf203c1be289f335052e98d54b8115713","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 17:24:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/09/668209/1/check/openstack-tox-linters/379dd1a/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7db1cb07e8e9825d355de7bf74737bad551d5d33","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 17:25:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/c0f9ffa/ : FAILURE in 8m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2420a17a865e12090de79a95437a7c61e86a75f1","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-06-28 18:52:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-354fc649bdb5aa084a31394cb9fec612a7a3c04b : FAILURE in 1h 36m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ff57550600593bc8fc3d6fc7ef5bec28db3a3e9b","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-01 08:44:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2f309f61b18c02149e940132bcfacb750b494721","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 08:50:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/09/668209/2/check/openstack-tox-linters/9fcb906/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"0d0bb5f0d141a4661fa4ef805d44b1c45e22117a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 08:52:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/1b0c522/ : FAILURE in 6m 53s","accounts_in_message":[],"_revision_number":2},{"id":"630bc000039037c8c8943adb93b78bc9183da7d4","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-01 09:50:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-5b924e5b4e433d9aa6e4e833507bd08d291c46ce : FAILURE in 1h 05m 24s","accounts_in_message":[],"_revision_number":2},{"id":"63fdd61126c96f73f89e533e8f18c0a2acaa8754","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-01 16:14:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa8b5f6a37d04e49724ce9df259a5af836d860a1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 16:25:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/73d72dd/ : FAILURE in 10m 28s","accounts_in_message":[],"_revision_number":3},{"id":"01b74a487d764d1b1e3c876a9b84ffa916cf39c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 16:28:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/09/668209/3/check/openstack-tox-linters/127baab/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":3},{"id":"ca2ee451903e75116dcb0da7adf409b51ddb51fd","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-01 16:39:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-326c7bb7a448401520a9bd8fda28078f1a9fbfd8 : FAILURE in 25m 15s","accounts_in_message":[],"_revision_number":3},{"id":"fb6668c14cadfa7b11b53fd0a2ef60a13c4fe39b","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-02 04:04:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a14012d98187ae417a8c8769f1c52fee84cdfdd2","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-02 07:48:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ae4b6927f6ed4b710e8559cafb5e415223da99e1","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-02 07:50:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-6a9c64ab7958fed27d0e466f416650766622756c : FAILURE in 2m 00s","accounts_in_message":[],"_revision_number":4},{"id":"e131f7b5d7eea9ad7c280fb39d69fdc0783afa39","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-02 07:56:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/8c7eb50/ : FAILURE in 7m 07s","accounts_in_message":[],"_revision_number":4},{"id":"c3e6bde8e8b18e4e3d4a682d54b0400f0fd4a59d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 07:58:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters http://logs.openstack.org/09/668209/4/check/openstack-tox-linters/013be54/ : FAILURE in 3m 10s","accounts_in_message":[],"_revision_number":4},{"id":"120328d852d1ef53331c400d12f4d768b3f2f4c5","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-02 16:15:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"674051d47fa8e6b984d0404d93171151c9841fbf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-02 16:23:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/5/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/82b414d/ : FAILURE in 8m 06s","accounts_in_message":[],"_revision_number":5},{"id":"848d01f943493ca3765da70861cd663661fb50ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 16:24:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/09/668209/5/check/openstack-tox-linters/c8b2318/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":5},{"id":"fd91629f7004b1ef97c86b1b888a5b3c50a154ae","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-02 16:31:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Master-83f91cb0c80a893342e2fa7d79b3c03d2098b043 : FAILURE in 15m 49s","accounts_in_message":[],"_revision_number":5},{"id":"aae97f789c845ea36e34a3bc69d29171d222be6d","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-14 15:48:10.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"0fc111de979c95fd56a148e324436aa60552027e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 15:56:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/09/668209/6/check/openstack-tox-linters/3ac2af6/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":6},{"id":"f95660484eb0ba49512a216e3ef3eb9aff013c26","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-14 15:57:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/6/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/4d823e0/ : FAILURE in 8m 41s","accounts_in_message":[],"_revision_number":6},{"id":"934318d22dc31c419b22700564341a36f923436c","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-14 16:07:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-e8752cbfddd3e497bd6cb3e087b5bbf9d1a3b39f : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":6},{"id":"9f7d97c81c47a36bd4758247a975f48a0bfdf5e3","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-23 19:37:05.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"19a3b78f5cb07a76043ecc5dc9e16b107fc937f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 19:50:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/09/668209/6/check/openstack-tox-linters/4b0e999/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":6},{"id":"f25929bfffcaa639b8d19282215ed6d1517a12f1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 19:54:19.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/6/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/48b2cb0/ : FAILURE in 16m 17s","accounts_in_message":[],"_revision_number":6},{"id":"63670c15078a0b198ae23cfb038d39e13191a1f6","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-24 01:31:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-f6dffb95628e93d8705de966bf434b54e4718584 : SUCCESS in 5h 54m 00s","accounts_in_message":[],"_revision_number":6},{"id":"6dd86ee7202580955a12309a89146f1075fd0b98","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-01 15:55:12.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"38797a68c37df48b4708f1f43da2a79d8eeca629","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 16:02:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/09/668209/7/check/openstack-tox-linters/a17900c/ : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":7},{"id":"a66e86083d6fe9a74a97d8665cff156dd264c687","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 16:07:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/7/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/6794cf8/ : FAILURE in 11m 57s","accounts_in_message":[],"_revision_number":7},{"id":"a96ee4ccc13a65716b0e794913f4c57080a5f355","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 16:14:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-00de62c32fea452954ac53680cedd1d64cbd9c4c : SUCCESS in 19m 10s","accounts_in_message":[],"_revision_number":7},{"id":"4f336e1f2b7bc87d3c7bcc14d1db8114ee17f425","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-01 17:17:40.000000000","message":"Patch Set 7:\n\nI thought PBR_VERSION is what\u0027s supposed to fix the \"fatal: not a git repository (or any of the parent directories): .git\" problem that\u0027s in the RDO CI, for some reason maybe the sphinx-build-3 command is ignoring it?","accounts_in_message":[],"_revision_number":7},{"id":"b63f273e3d4fb0c9ae99544f60652b6b3960443c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-02 14:21:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"48ba98029a14b4cce3164df46d08c8126944fedd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 14:31:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/09/668209/8/check/openstack-tox-linters/ad3abe9/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":8},{"id":"1691b0bafe9295809ad64fcaa77f09f543e65f57","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 14:33:45.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/8/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/7f7c788/ : FAILURE in 12m 08s","accounts_in_message":[],"_revision_number":8},{"id":"5e3b061da837cb540a3a71d61e5e96b7f2ef0c50","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 14:38:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-1d6b03d88b16eb5be720b807de96f5274732efd6 : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":8},{"id":"fb60a3793a230352568ff1107c4e8dd2bea27c41","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-02 20:22:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e25756ceeed45b6cb88eff5d2827f33c3a163541","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 20:31:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://logs.opendev.org/09/668209/9/check/openstack-tox-linters/f22c857/ : FAILURE in 3m 49s","accounts_in_message":[],"_revision_number":9},{"id":"8e2e47fba94d32f351e70f8c04792796275e67f5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 20:35:09.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/9/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/deb185e/ : FAILURE in 11m 59s","accounts_in_message":[],"_revision_number":9},{"id":"603de02343c424a1493aa8117b3f141f68b72848","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-02 20:42:57.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"45ffb16e96c2d55109b3e0da1a84e9753860d102","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 20:52:06.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/09/668209/10/check/openstack-tox-linters/490dfe1/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":10},{"id":"8051e799d2e73a5ae22a816f81ab82c045be4841","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 20:55:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/10/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/d17c74b/ : FAILURE in 11m 21s","accounts_in_message":[],"_revision_number":10},{"id":"c933089ba3b490c1a824c00a0ca994c44d3c0152","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-02 20:59:40.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"314df6f02dade68134cb15e0d4ba6628dd8755e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 21:07:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/09/668209/11/check/openstack-tox-linters/1dacaf6/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":11},{"id":"1231320984f2172bdfc62d037262bb74ba900261","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 21:18:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/11/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/d174559/ : FAILURE in 18m 31s","accounts_in_message":[],"_revision_number":11},{"id":"990d1301418cb8b28ef76522562d1ea9f3a10e4b","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 22:10:38.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-1208f089ff9354458ddb85469e8c89656b5be470 : SUCCESS in 15m 42s","accounts_in_message":[],"_revision_number":11},{"id":"b0d804530ef48aadf37a9de03a746d6e2fc069a8","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-04 17:52:54.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3d41aa05de6e8ef4fc6032c65b0816c4e6b295be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 17:57:16.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/09/668209/12/check/openstack-tox-linters/bf92b2e/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":12},{"id":"bf05398f454813c92038004730a871772b47d818","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-04 18:21:14.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-f36cc9b4afda6ff74b950336371f23a3f3e1a059 : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":12},{"id":"cae6db9fc9e9f7570eae00d2deb28cf572aeb7ff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-04 18:22:10.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/12/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/da6e087/ : FAILURE in 28m 32s","accounts_in_message":[],"_revision_number":12},{"id":"b7d0fb7df29c00e5db6cc26bf36e8c86e6650956","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-12 14:30:29.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"59f23f3b4576590b540962d3bab603faf5290c2e","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-13 04:35:48.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d6fede218ef8c388f9f47d935263e2defcc42539","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-13 15:35:50.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"e87a986242a265cc48fc6f130f43e6886f33cb7a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-13 15:36:21.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"19575d27da2b442bbf96889e58e09fcb29819a4e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-13 15:38:53.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"acf49f539d680df133a22cf24c5321ddcaf22939","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-13 15:41:35.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8ae2ea8bc6d940a30e8be769b3737263db0ddddc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 15:48:14.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/139682ca449c482884e920a4106fdfc6 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":13},{"id":"2a5062c52930b1b7f907763608732911775451a2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 16:03:19.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/13/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/7fc63ff/ : FAILURE in 23m 54s","accounts_in_message":[],"_revision_number":13},{"id":"887b24871a392313d1909e639d7f159495551add","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-13 19:40:55.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"015bac81738d61a02f5dd0fbf14d3c2a0be8932f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 19:53:10.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/486673f798ee4da8a97daf11fd55070f : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":14},{"id":"968052f6f41d1e9d72aa792a7d773f506cabfb66","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 19:59:33.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/09/668209/14/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/f4ec5cc/ : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":14},{"id":"9037fb6f5e2b0fe931f2bc3644fc2d5e02e21a41","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-13 20:00:19.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-780bc293f1ec032d707a455ba7d00e6604807886 : SUCCESS in 19m 04s","accounts_in_message":[],"_revision_number":14},{"id":"ef6747f5fe7983e423b4e21d8dbf18ca04968873","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-14 07:31:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e9816f71b91806ff32af5068e0d0b70f8a5c4e34","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-21 07:16:48.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"2adb5154c2216d2ec3caaff94097e315920e2d67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:17:49.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"5e31d78075b13ef3da0a5b677df1dfdf054ca916","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:27:51.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/efd82f26d8964c7e926930c035912b33 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":14},{"id":"b090bbfe345cbd0f6878c4d1ecacc4371671d944","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:27:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"780bc293f1ec032d707a455ba7d00e6604807886","revisions":{"354fc649bdb5aa084a31394cb9fec612a7a3c04b":{"kind":"REWORK","_number":1,"created":"2019-06-28 17:15:42.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/1"}}},"commit":{"parents":[{"commit":"d2406bf5fe8b01ecd9f02df7847bec235ef27c6c","subject":"Add sphinx_build macro for better redhat/suse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d2406bf5fe8b01ecd9f02df7847bec235ef27c6c"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:14:51.000000000","tz":120},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/354fc649bdb5aa084a31394cb9fec612a7a3c04b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/354fc649bdb5aa084a31394cb9fec612a7a3c04b"}]},"branch":"refs/heads/master"},"5b924e5b4e433d9aa6e4e833507bd08d291c46ce":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-01 08:44:51.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/2"}}},"commit":{"parents":[{"commit":"089f4c4ff12a78028e4b56b624d8530dc4822375","subject":"Merge \"Fix build for masakari-monitors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/089f4c4ff12a78028e4b56b624d8530dc4822375"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-01 08:44:51.000000000","tz":0},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5b924e5b4e433d9aa6e4e833507bd08d291c46ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5b924e5b4e433d9aa6e4e833507bd08d291c46ce"}]},"branch":"refs/heads/master"},"326c7bb7a448401520a9bd8fda28078f1a9fbfd8":{"kind":"REWORK","_number":3,"created":"2019-07-01 16:14:16.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/3"}}},"commit":{"parents":[{"commit":"089f4c4ff12a78028e4b56b624d8530dc4822375","subject":"Merge \"Fix build for masakari-monitors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/089f4c4ff12a78028e4b56b624d8530dc4822375"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-01 16:14:07.000000000","tz":120},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/326c7bb7a448401520a9bd8fda28078f1a9fbfd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/326c7bb7a448401520a9bd8fda28078f1a9fbfd8"}]},"branch":"refs/heads/master"},"6a9c64ab7958fed27d0e466f416650766622756c":{"kind":"REWORK","_number":4,"created":"2019-07-02 07:48:30.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/4"}}},"commit":{"parents":[{"commit":"089f4c4ff12a78028e4b56b624d8530dc4822375","subject":"Merge \"Fix build for masakari-monitors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/089f4c4ff12a78028e4b56b624d8530dc4822375"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-02 07:48:18.000000000","tz":120},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6a9c64ab7958fed27d0e466f416650766622756c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6a9c64ab7958fed27d0e466f416650766622756c"}]},"branch":"refs/heads/master"},"83f91cb0c80a893342e2fa7d79b3c03d2098b043":{"kind":"REWORK","_number":5,"created":"2019-07-02 16:15:13.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/5"}}},"commit":{"parents":[{"commit":"089f4c4ff12a78028e4b56b624d8530dc4822375","subject":"Merge \"Fix build for masakari-monitors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/089f4c4ff12a78028e4b56b624d8530dc4822375"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-02 16:15:03.000000000","tz":120},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/83f91cb0c80a893342e2fa7d79b3c03d2098b043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/83f91cb0c80a893342e2fa7d79b3c03d2098b043"}]},"branch":"refs/heads/master"},"e8752cbfddd3e497bd6cb3e087b5bbf9d1a3b39f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-07-14 15:48:10.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/6"}}},"commit":{"parents":[{"commit":"2fbca8bf654f17b1d591edbbd1eee5a146b46d50","subject":"os-xenapi: switch to singlespec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/2fbca8bf654f17b1d591edbbd1eee5a146b46d50"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-14 15:48:10.000000000","tz":0},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/e8752cbfddd3e497bd6cb3e087b5bbf9d1a3b39f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/e8752cbfddd3e497bd6cb3e087b5bbf9d1a3b39f"}]},"branch":"refs/heads/master"},"00de62c32fea452954ac53680cedd1d64cbd9c4c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-01 15:55:12.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/7"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-01 15:55:12.000000000","tz":0},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/00de62c32fea452954ac53680cedd1d64cbd9c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/00de62c32fea452954ac53680cedd1d64cbd9c4c"}]},"branch":"refs/heads/master"},"ecd5db8ce1e6299e940b1c9e1914254122428fa9":{"kind":"REWORK","_number":8,"created":"2019-08-02 14:21:05.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/8"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-02 14:20:56.000000000","tz":120},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ecd5db8ce1e6299e940b1c9e1914254122428fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ecd5db8ce1e6299e940b1c9e1914254122428fa9"}]},"branch":"refs/heads/master"},"b01f0a9d442ff1dae8d08c3b4c8d55dd7bf31764":{"kind":"REWORK","_number":9,"created":"2019-08-02 20:22:09.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/9"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-02 20:21:56.000000000","tz":120},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b01f0a9d442ff1dae8d08c3b4c8d55dd7bf31764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b01f0a9d442ff1dae8d08c3b4c8d55dd7bf31764"}]},"branch":"refs/heads/master"},"822a7428794360d2efb738f9056371be7e7258db":{"kind":"REWORK","_number":10,"created":"2019-08-02 20:42:57.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/10"}}},"commit":{"parents":[{"commit":"b9a6ec6f782cfb8bd67e4a56d53a949b6b4262ee","subject":"Merge \"nova: Remove nova-consoleauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b9a6ec6f782cfb8bd67e4a56d53a949b6b4262ee"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-02 20:42:48.000000000","tz":120},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/822a7428794360d2efb738f9056371be7e7258db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/822a7428794360d2efb738f9056371be7e7258db"}]},"branch":"refs/heads/master"},"1208f089ff9354458ddb85469e8c89656b5be470":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-08-02 20:59:40.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/11"}}},"commit":{"parents":[{"commit":"b914ead3b75cc9dc91a1c482e6f1771a476c8330","subject":"openstackdocstheme: update to 1.31.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b914ead3b75cc9dc91a1c482e6f1771a476c8330"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-02 20:59:40.000000000","tz":0},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1208f089ff9354458ddb85469e8c89656b5be470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1208f089ff9354458ddb85469e8c89656b5be470"}]},"branch":"refs/heads/master"},"f36cc9b4afda6ff74b950336371f23a3f3e1a059":{"kind":"REWORK","_number":12,"created":"2019-08-04 17:52:54.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/12"}}},"commit":{"parents":[{"commit":"b914ead3b75cc9dc91a1c482e6f1771a476c8330","subject":"openstackdocstheme: update to 1.31.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b914ead3b75cc9dc91a1c482e6f1771a476c8330"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-04 17:52:42.000000000","tz":120},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/f36cc9b4afda6ff74b950336371f23a3f3e1a059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/f36cc9b4afda6ff74b950336371f23a3f3e1a059"}]},"branch":"refs/heads/master"},"9f265606c16f8d924ff89590945d3c8d05e2c1df":{"kind":"REWORK","_number":13,"created":"2019-08-13 15:38:53.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/13"}}},"commit":{"parents":[{"commit":"1d3a778f27471832f2e3aa7a01e6e69c8bc2d7d9","subject":"Fix provides/requires for ironic-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1d3a778f27471832f2e3aa7a01e6e69c8bc2d7d9"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-13 15:38:43.000000000","tz":120},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9f265606c16f8d924ff89590945d3c8d05e2c1df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9f265606c16f8d924ff89590945d3c8d05e2c1df"}]},"branch":"refs/heads/master"},"780bc293f1ec032d707a455ba7d00e6604807886":{"kind":"REWORK","_number":14,"created":"2019-08-13 19:40:55.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/09/668209/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/09/668209/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/09/668209/14"}}},"commit":{"parents":[{"commit":"1d3a778f27471832f2e3aa7a01e6e69c8bc2d7d9","subject":"Fix provides/requires for ironic-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1d3a778f27471832f2e3aa7a01e6e69c8bc2d7d9"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 17:13:00.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-13 19:40:47.000000000","tz":120},"subject":"Switch to python3 for cinder","message":"Switch to python3 for cinder\n\nIt\u0027s becoming more bothersome to keep sphinx-build and oslo work\nwith python 2 and python 3 at the same time, so lets take the\nbullet and switch it to python 3, which fixes the build.\n\nChange-Id: I7a9d8a8cbe508cee446fa1b7f512f510c2187c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/780bc293f1ec032d707a455ba7d00e6604807886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/780bc293f1ec032d707a455ba7d00e6604807886"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
