)]}'
{"id":"openstack%2Frpm-packaging~670357","triplet_id":"openstack%2Frpm-packaging~master~Icb00b38b639da7b3f20b472fb39efbc72b44f81c","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"Icb00b38b639da7b3f20b472fb39efbc72b44f81c","subject":"neutron/networking-generic-switch: switch to python3","status":"MERGED","created":"2019-07-11 17:20:18.000000000","updated":"2019-08-08 15:04:50.000000000","submitted":"2019-08-08 15:04:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670357-1565276690908-3f7c88c0","meta_rev_id":"0b48b3c5554d3a7b36949994e9a214aa9573431e","_number":670357,"virtual_id_number":670357,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-08 14:23:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-08-08 15:04:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-08 13:36:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-08-08 14:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-08 13:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-08-08 14:54:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 06:54:59.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-08-08 13:35:39.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-08-08 13:36:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-08 14:23:16.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-08 14:54:28.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-08-08 15:04:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66f308d4881366b25d80785da32453f936fb556d","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-11 17:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27972a3a6661ab8bb182876f195b4050c41cef5c","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-11 17:24:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-8a4ce11536245bdd3543bce8308a5fbe7fa936b9 : FAILURE in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"68fedf6680c4d396221ca22873d1d6adbad66736","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 17:28:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters http://logs.openstack.org/57/670357/1/check/openstack-tox-linters/d418c6c/ : FAILURE in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"06a4d074a46ce63103b79a30efd15bbe8790a9f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 17:30:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/e3e182c/ : FAILURE in 9m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d44503861937be5a87911ceb646d86427f9cebc1","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-11 17:40:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cea44547871a922c8039d8bc1395a1de45f866e4","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-11 17:44:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-ffd22b3f3d557457af07a4062d40dc6cbe93acb5 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"5a5f32dae4766f9e2bee5fa2b016a511174a1a09","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 17:47:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/0e59513/ : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":2},{"id":"45ebb80accd798554dc1dce36ff50d28ffa95b9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 17:52:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters http://logs.openstack.org/57/670357/2/check/openstack-tox-linters/96e1a61/ : FAILURE in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"22200403e9778787427ffc057679eafd43eff802","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-11 20:39:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1478a3930cfe6ecfefe790e6f3f403e01aafa3ba","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-11 20:43:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-3e533c0d6e654fbe44aa3e5545e09b908998f247 : FAILURE in 4m 26s","accounts_in_message":[],"_revision_number":3},{"id":"17a04fc0e11a7a2b8a9f97ee4e66f8c25b6b15f5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 20:49:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/737f521/ : FAILURE in 10m 19s","accounts_in_message":[],"_revision_number":3},{"id":"e51b68b5f3d68e0e18c79134fa7f44d99c5952f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 20:50:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters http://logs.openstack.org/57/670357/3/check/openstack-tox-linters/e0ef3c9/ : FAILURE in 3m 52s","accounts_in_message":[],"_revision_number":3},{"id":"46ad4f61a71d2a617f374307dfe57e2da57a3004","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-11 21:16:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8aff8c53d4dc0af438e489bdc062842401ee07f8","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-11 21:21:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-939fb309610417fd77a5831982d16b6d026b194e : FAILURE in 4m 22s","accounts_in_message":[],"_revision_number":4},{"id":"4d212ae5e2ac689160b982287a647cbfb9f0b4d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 21:26:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/57/670357/4/check/openstack-tox-linters/60aace4/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":4},{"id":"b726241f067b4a1beac8c8ab5512177db90a61d6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 21:26:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/19333a8/ : FAILURE in 10m 00s","accounts_in_message":[],"_revision_number":4},{"id":"5a692b984a0b04f11a62f676069f7fc282262f2a","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-12 17:00:23.000000000","message":"Patch Set 4:\n\nrecheck\n\nsuse CI is not using the right version of renderspec https://ci.opensuse.org/job/openstack-rpm-packaging-sles15-Master/36/console","accounts_in_message":[],"_revision_number":4},{"id":"4e501d197d200f4ac10519ed5071ecdfbfef4973","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-12 17:06:06.000000000","message":"Patch Set 4:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-939fb309610417fd77a5831982d16b6d026b194e : FAILURE in 5m 24s","accounts_in_message":[],"_revision_number":4},{"id":"89d0235b7a43f1c8bfee37a3fc3b04e391327593","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-12 17:07:10.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/ec4284f/ : FAILURE in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"a152e074706a61d91c4942c433f2d57215e3d7f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 17:11:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/57/670357/4/check/openstack-tox-linters/6ff86db/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":4},{"id":"0bc98e1a8ca9780a1c99562a47f121e501d2c492","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-15 07:00:51.000000000","message":"Patch Set 4:\n\nyeah, I need to upgrade the workers to SLE15, didn\u0027t have time for that yet ;-/","accounts_in_message":[],"_revision_number":4},{"id":"c63b2d3b2b43941214eea736dd178383e4b504a5","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-15 10:35:51.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"39f8f3cc5a03b1253de7fe98c2a43ed10d311ac6","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 10:36:13.000000000","message":"Patch Set 4:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-0d48810941fb126b7f00566dbb9681b856b49069 : FAILURE in 3s","accounts_in_message":[],"_revision_number":4},{"id":"7e1cda5d47ce46eed2c54635be074f11e874badd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 10:41:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/57/670357/4/check/openstack-tox-linters/4d09fb7/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":4},{"id":"87e8ce4744ca77bc1c9a4286bb188e592822a65d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 10:44:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/dec1c18/ : FAILURE in 7m 18s","accounts_in_message":[],"_revision_number":4},{"id":"27eea8f7ac66dcdf234c7aa37c497760d18b0e20","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-15 15:38:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6b0daf748b1f942a055672443dcdaf32e755b28e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 15:46:29.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/6d62811/ : FAILURE in 6m 54s","accounts_in_message":[],"_revision_number":4},{"id":"fc84bf134b9c6e81b5a17d412f8e739f289ec3f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 15:50:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/57/670357/4/check/openstack-tox-linters/cd60a76/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":4},{"id":"a2348f40e8c2806562e235693835935c4892129f","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 16:38:18.000000000","message":"Patch Set 4:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://ci.opensuse.org/logs/670357/4/rpm-packaging/openstack-rpm-packaging-sles15-Master/56 : ABORTED in 59m 09s","accounts_in_message":[],"_revision_number":4},{"id":"24bd6a471136c47a6ec38cbe92f2c0a5dc44ed37","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-15 17:02:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"86b54ccf2a9899566608588216304d60f13e31d2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 17:09:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/244fd0b/ : FAILURE in 6m 00s","accounts_in_message":[],"_revision_number":4},{"id":"7bc938a7b5074a5a379bd5e8a5713daad35ba680","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 17:09:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/57/670357/4/check/openstack-tox-linters/f202766/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":4},{"id":"776d3550a90b004bbc48eb556e9ecd3e13285f72","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 17:17:42.000000000","message":"Patch Set 4:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-c5cd29037c35dc082fa75e9e87217c9cdc1598c1 : FAILURE in 14m 56s","accounts_in_message":[],"_revision_number":4},{"id":"7a6630ca32e0e45864a29d6421e8c7ab2faefee2","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-07-22 15:18:34.000000000","message":"Patch Set 4:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":4},{"id":"c22c26aa5fae56fec0a7f2d52b555f4accf028ac","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 15:28:10.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/3167951/ : FAILURE in 9m 06s","accounts_in_message":[],"_revision_number":4},{"id":"3f433df1cbcfa0a0294a70bf309bdf2feb855efa","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-22 21:22:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"dbee8e955b95feba70b20b36e9412dfe335286ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 21:28:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/57/670357/4/check/openstack-tox-linters/ae8f45f/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"6516ab6e2f73718893b9f377cd4b4ad6f169b567","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 21:31:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/59a1276/ : FAILURE in 8m 46s","accounts_in_message":[],"_revision_number":4},{"id":"6ffb89e177c20c94f55f60ab8bd1d99ade6623dc","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-22 22:11:03.000000000","message":"Patch Set 4:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-78de91586b212f0acb8a8f4a3d6f00fddb7fcd03 : FAILURE in 48m 24s","accounts_in_message":[],"_revision_number":4},{"id":"008f5efe0127fc3db864c97e3244bb609bf1ccaa","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-24 20:17:29.000000000","message":"Patch Set 4: Code-Review+2\n\nSeems that this relies on https://review.opendev.org/671455 and vice versa","accounts_in_message":[],"_revision_number":4},{"id":"535776892258e6727aa950ee018d4a87c48678bd","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-07-25 06:54:59.000000000","message":"Patch Set 4:\n\nThe RDO CI is failing with:\n\nDEBUG: BUILDSTDERR: fatal: not a git repository (or any of the parent directories): .git\n\nWhat was the fix for that?","accounts_in_message":[],"_revision_number":4},{"id":"21f0028d7c8985ad1cf24074813a4dc6d11fc763","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-02 14:28:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"59ab905fec521be3f2640a754f7ae1221055aa07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 14:36:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/57/670357/5/check/openstack-tox-linters/03e89ce/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":5},{"id":"11d46f3717d5d1c85afa92ddcceebaccd5c9fe80","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 14:37:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/5/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/888c4f5/ : FAILURE in 8m 39s","accounts_in_message":[],"_revision_number":5},{"id":"3e55da96846a4abb95a88567eec98e0e0c85325a","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 14:46:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-ba70af6a4d9d71dfb8865dbae999b5e2fe3688e0 : FAILURE in 18m 04s","accounts_in_message":[],"_revision_number":5},{"id":"0a04c7072a63448d956f0684398c346c819b9569","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-02 20:52:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b30d2a3567b9e635c25542396748d267c5ae45de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 21:01:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/57/670357/6/check/openstack-tox-linters/ee14528/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":6},{"id":"1d074e05d793554dba09a3a96acc13e27fa54a94","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 21:02:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/6/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/cdccd8c/ : FAILURE in 9m 28s","accounts_in_message":[],"_revision_number":6},{"id":"03b9dd97e1fde82daa02f31131cbf8efd550e862","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 23:08:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-3a306b33a84629eea52067e3a4f646a21f785aad : FAILURE in 2h 08m 34s","accounts_in_message":[],"_revision_number":6},{"id":"f5ffce08a1ee3140ff87998efe23a5fb5495058a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-04 17:54:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b136bc4026eb380a894e99fd9b15bc92ad8781ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 17:58:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/57/670357/7/check/openstack-tox-linters/f6cc612/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":7},{"id":"5e2d86fda5ee867098ff0f76285c2e4b5f2be8ad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-04 18:04:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/7/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/06930e8/ : FAILURE in 9m 54s","accounts_in_message":[],"_revision_number":7},{"id":"d44de5caac4954b66f5e08f57c944ecb610f5d1e","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-04 18:22:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-9972daaccbfcb73b4f43d7185291406241f44413 : SUCCESS in 27m 57s","accounts_in_message":[],"_revision_number":7},{"id":"510b77fcaf8c704a921434055c558af5269a1974","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-05 19:01:03.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"3621cabd249592ce44cdb7523efd48b231ced37f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 19:10:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/57/670357/7/check/openstack-tox-linters/1878a8d/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":7},{"id":"7e74c429f20b412cee1501f0d12b5e43e427c457","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 19:11:41.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/7/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/18adc9b/ : FAILURE in 9m 14s","accounts_in_message":[],"_revision_number":7},{"id":"b519cb59c828ba41346f28870302b80711b907cb","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-05 19:40:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-5ea3647cf87c98189ef07a08b0a94385d7d56155 : FAILURE in 39m 27s","accounts_in_message":[],"_revision_number":7},{"id":"019527a24e2c997f8d4c9adf6aec1a7ce6f2fd1d","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-06 10:40:30.000000000","message":"Patch Set 7:\n\nSUSE CI needs a netmiko update to 2.4.1 . Done in dlp and dlpb","accounts_in_message":[],"_revision_number":7},{"id":"8776e304c173b30defcd90d5dc48c44b99ec1579","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-08 13:26:35.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"4ab08c81656ba27def1a8cd459e29e4d62482fac","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-08 13:35:39.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"82911ad75a9654172397ddb957e36581e94dd3bd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 13:36:45.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/57/670357/8/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/edafef5/ : FAILURE in 9m 38s","accounts_in_message":[],"_revision_number":8},{"id":"457be2734c78acb0709e514f9066ff5012c8b6e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 13:36:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/57/670357/8/check/openstack-tox-linters/a370796/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":8},{"id":"6b67428c3a4aad5e4104c99daaf3fb1505130949","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 14:23:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-77f2fcb936a7da56239cdbe52e5f1154184f8f48 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":8},{"id":"56614f9a308714fb3b711a76f9a8579eacb45827","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-08 14:54:28.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"a5abe15539ce341190c831a756fed3d7505055e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:54:41.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c52436db9471706d15b2db11d308edf7a8c214df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:04:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"0b48b3c5554d3a7b36949994e9a214aa9573431e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:04:50.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/57/670357/8/gate/openstack-tox-linters/ca92b8b/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"77f2fcb936a7da56239cdbe52e5f1154184f8f48","revisions":{"8a4ce11536245bdd3543bce8308a5fbe7fa936b9":{"kind":"REWORK","_number":1,"created":"2019-07-11 17:20:18.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/57/670357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/57/670357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/1"}}},"commit":{"parents":[{"commit":"9a313cfbccad9ffdfc89abe4e936ce87248d6d5d","subject":"Merge \"oslo.config update to 6.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9a313cfbccad9ffdfc89abe4e936ce87248d6d5d"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-11 17:18:59.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-11 17:19:54.000000000","tz":120},"subject":"neutron: switch to python3","message":"neutron: switch to python3\n\nChange-Id: Icb00b38b639da7b3f20b472fb39efbc72b44f81c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8a4ce11536245bdd3543bce8308a5fbe7fa936b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8a4ce11536245bdd3543bce8308a5fbe7fa936b9"}]},"branch":"refs/heads/master"},"ffd22b3f3d557457af07a4062d40dc6cbe93acb5":{"kind":"REWORK","_number":2,"created":"2019-07-11 17:40:13.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/57/670357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/57/670357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/2"}}},"commit":{"parents":[{"commit":"ab1bbfd8ca95f165a5c5f7dfde8c6e97ca720b88","subject":"os-xenapi: switch to singlespec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ab1bbfd8ca95f165a5c5f7dfde8c6e97ca720b88"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-11 17:18:59.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-11 17:40:03.000000000","tz":120},"subject":"neutron: switch to python3","message":"neutron: switch to python3\n\nChange-Id: Icb00b38b639da7b3f20b472fb39efbc72b44f81c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ffd22b3f3d557457af07a4062d40dc6cbe93acb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ffd22b3f3d557457af07a4062d40dc6cbe93acb5"}]},"branch":"refs/heads/master"},"3e533c0d6e654fbe44aa3e5545e09b908998f247":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-11 20:39:05.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/57/670357/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/57/670357/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/3"}}},"commit":{"parents":[{"commit":"2fbca8bf654f17b1d591edbbd1eee5a146b46d50","subject":"os-xenapi: switch to singlespec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/2fbca8bf654f17b1d591edbbd1eee5a146b46d50"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-11 17:18:59.000000000","tz":120},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-07-11 20:39:05.000000000","tz":0},"subject":"neutron: switch to python3","message":"neutron: switch to python3\n\nChange-Id: Icb00b38b639da7b3f20b472fb39efbc72b44f81c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3e533c0d6e654fbe44aa3e5545e09b908998f247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3e533c0d6e654fbe44aa3e5545e09b908998f247"}]},"branch":"refs/heads/master"},"939fb309610417fd77a5831982d16b6d026b194e":{"kind":"REWORK","_number":4,"created":"2019-07-11 21:16:23.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/57/670357/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/57/670357/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/4"}}},"commit":{"parents":[{"commit":"2fbca8bf654f17b1d591edbbd1eee5a146b46d50","subject":"os-xenapi: switch to singlespec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/2fbca8bf654f17b1d591edbbd1eee5a146b46d50"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-11 17:18:59.000000000","tz":120},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-11 21:16:17.000000000","tz":-420},"subject":"neutron: switch to python3","message":"neutron: switch to python3\n\nChange-Id: Icb00b38b639da7b3f20b472fb39efbc72b44f81c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/939fb309610417fd77a5831982d16b6d026b194e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/939fb309610417fd77a5831982d16b6d026b194e"}]},"branch":"refs/heads/master"},"21a6ce4c8450f33771e39e502f437d771d4f8174":{"kind":"REWORK","_number":5,"created":"2019-08-02 14:28:32.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/57/670357/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/57/670357/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/5"}}},"commit":{"parents":[{"commit":"2fbca8bf654f17b1d591edbbd1eee5a146b46d50","subject":"os-xenapi: switch to singlespec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/2fbca8bf654f17b1d591edbbd1eee5a146b46d50"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-11 17:18:59.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-02 14:28:24.000000000","tz":120},"subject":"neutron: switch to python3","message":"neutron: switch to python3\n\nChange-Id: Icb00b38b639da7b3f20b472fb39efbc72b44f81c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/21a6ce4c8450f33771e39e502f437d771d4f8174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/21a6ce4c8450f33771e39e502f437d771d4f8174"}]},"branch":"refs/heads/master"},"3a306b33a84629eea52067e3a4f646a21f785aad":{"kind":"REWORK","_number":6,"created":"2019-08-02 20:52:39.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/57/670357/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/57/670357/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/6"}}},"commit":{"parents":[{"commit":"b9a6ec6f782cfb8bd67e4a56d53a949b6b4262ee","subject":"Merge \"nova: Remove nova-consoleauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b9a6ec6f782cfb8bd67e4a56d53a949b6b4262ee"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-11 17:18:59.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-02 20:51:53.000000000","tz":120},"subject":"neutron/networking-generic-switch: switch to python3","message":"neutron/networking-generic-switch: switch to python3\n\nWe need both to merge at the same time as they\ndepend on each other.\n\nCo-Authored-By: Jaime Caamaño Ruiz \u003cjcaamano@suse.com\u003e\n\nChange-Id: Icb00b38b639da7b3f20b472fb39efbc72b44f81c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3a306b33a84629eea52067e3a4f646a21f785aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3a306b33a84629eea52067e3a4f646a21f785aad"}]},"branch":"refs/heads/master"},"a593d6a83d341f27bd7900758be090d764e396a6":{"kind":"REWORK","_number":7,"created":"2019-08-04 17:54:01.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/57/670357/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/57/670357/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/7"}}},"commit":{"parents":[{"commit":"b9a6ec6f782cfb8bd67e4a56d53a949b6b4262ee","subject":"Merge \"nova: Remove nova-consoleauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b9a6ec6f782cfb8bd67e4a56d53a949b6b4262ee"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-11 17:18:59.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-04 17:53:53.000000000","tz":120},"subject":"neutron/networking-generic-switch: switch to python3","message":"neutron/networking-generic-switch: switch to python3\n\nWe need both to merge at the same time as they\ndepend on each other.\n\nCo-Authored-By: Jaime Caamaño Ruiz \u003cjcaamano@suse.com\u003e\n\nChange-Id: Icb00b38b639da7b3f20b472fb39efbc72b44f81c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a593d6a83d341f27bd7900758be090d764e396a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a593d6a83d341f27bd7900758be090d764e396a6"}]},"branch":"refs/heads/master"},"77f2fcb936a7da56239cdbe52e5f1154184f8f48":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-08 13:26:35.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/57/670357/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/57/670357/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/57/670357/8"}}},"commit":{"parents":[{"commit":"264ce687c73c6b0419ce8ed83eff06bde548d733","subject":"Merge \"monasca-persister: Switch to python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/264ce687c73c6b0419ce8ed83eff06bde548d733"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-11 17:18:59.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-08 13:26:35.000000000","tz":0},"subject":"neutron/networking-generic-switch: switch to python3","message":"neutron/networking-generic-switch: switch to python3\n\nWe need both to merge at the same time as they\ndepend on each other.\n\nCo-Authored-By: Jaime Caamaño Ruiz \u003cjcaamano@suse.com\u003e\n\nChange-Id: Icb00b38b639da7b3f20b472fb39efbc72b44f81c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/77f2fcb936a7da56239cdbe52e5f1154184f8f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/77f2fcb936a7da56239cdbe52e5f1154184f8f48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
