)]}'
{"id":"openstack%2Frpm-packaging~670602","triplet_id":"openstack%2Frpm-packaging~stable%2Fpike~I7d08f5f0c3d1445c249b2c0715cc0e435069293c","project":"openstack/rpm-packaging","branch":"stable/pike","hashtags":[],"change_id":"I7d08f5f0c3d1445c249b2c0715cc0e435069293c","subject":"monasca-agent: update filelist","status":"MERGED","created":"2019-07-12 16:57:07.000000000","updated":"2019-07-17 21:06:55.000000000","submitted":"2019-07-17 21:06:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670602-1563397615916-6d7173bf","meta_rev_id":"95d0b118d8ff8213c1c240cb590358eeb5389f92","_number":670602,"virtual_id_number":670602,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-07-12 17:14:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-07-17 21:06:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-12 17:11:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-07-17 16:03:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-12 16:59:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-17 20:57:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 16:59:04.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-07-12 17:11:56.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-12 17:14:00.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-17 16:03:22.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-07-17 21:06:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f568035f83c5fba719254f1f4897ad3898d8e894","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-12 16:57:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"937ddd26087d01bc59b8874c48f0995f5ef0ba95","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-12 16:59:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c094ba43482b1643e1ee6dc6f33ef7d7b6c155d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 17:09:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/02/670602/1/check/openstack-tox-linters/c5d87dd/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5214ccb7825e1e3ac1d1b4e665577438db8708c0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-12 17:11:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/02/670602/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/a69f162/ : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":1},{"id":"fdf973ec60f0e990286c85b05fdfd90be8be3cc7","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-12 17:14:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Pike https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Pike-cc6d597f895310c71195cbf1be0bc08e3f8c90c4 : SUCCESS in 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c46918c91c103db47a9803d15a892a6060ec8515","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-17 16:03:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb144c81b901270708bc401452f89c521ab44dcc","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-17 20:57:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"153305622f4334fb206d16d45cc312d9b7e90717","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 20:58:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7c1fdc8d57e6efc4a0239d5251a5d0d1c3fe912","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 21:06:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"95d0b118d8ff8213c1c240cb590358eeb5389f92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 21:06:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/02/670602/1/gate/openstack-tox-linters/5125999/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc6d597f895310c71195cbf1be0bc08e3f8c90c4","revisions":{"cc6d597f895310c71195cbf1be0bc08e3f8c90c4":{"kind":"REWORK","_number":1,"created":"2019-07-12 16:57:07.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/02/670602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/02/670602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/02/670602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/02/670602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/02/670602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/02/670602/1"}}},"commit":{"parents":[{"commit":"45b547c046983198d777a11d154e8d032f87f791","subject":"Don\u0027t exclude python bytecode files in dashboards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/45b547c046983198d777a11d154e8d032f87f791"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-12 16:56:54.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-12 16:56:54.000000000","tz":120},"subject":"monasca-agent: update filelist","message":"monasca-agent: update filelist\n\nChange-Id: I7d08f5f0c3d1445c249b2c0715cc0e435069293c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/cc6d597f895310c71195cbf1be0bc08e3f8c90c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/cc6d597f895310c71195cbf1be0bc08e3f8c90c4"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
