)]}'
{"id":"openstack%2Frpm-packaging~671976","triplet_id":"openstack%2Frpm-packaging~master~Ibb3cbfc6c940ba5a436fcb13e8d9200e647acc78","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"Ibb3cbfc6c940ba5a436fcb13e8d9200e647acc78","subject":"Install the pyc as well for the filelist capturing it","status":"MERGED","created":"2019-07-22 07:49:36.000000000","updated":"2020-07-31 13:40:08.000000000","submitted":"2020-07-31 13:40:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671976-1596202808380-2eb1796f","meta_rev_id":"ba8d54ab6074db6cabad988f6bbd05a79bf08949","_number":671976,"virtual_id_number":671976,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-09-13 09:29:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-31 13:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-09-11 18:32:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"recommended":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-11 18:17:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"value":2,"date":"2019-09-11 18:17:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"date":"2020-07-31 13:04:50.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-31 13:17:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 12:53:31.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-09-10 12:57:25.000000000","updated_by":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"reviewer":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"state":"REVIEWER"},{"updated":"2019-09-11 18:32:16.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-13 09:29:26.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-31 13:04:50.000000000","updated_by":{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"},"reviewer":{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"},"state":"REVIEWER"},{"updated":"2020-07-31 13:40:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5835bc19fc216ac48613f33e3a98982bcd5daf99","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-22 07:49:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34165b7fabf61e673d96d8ea507955b4da8abeb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 08:06:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/76/671976/1/check/openstack-tox-linters/5b8899b/ : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4a3ae55f6b992b217710ca61dae02cbc677e8f92","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-22 08:22:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-8d698d86b7b35516b4f89031ee194074dee75a47 : FAILURE in 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b6574a88cdb40ef10a07a07201934dcdcd05acf3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 10:02:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/76/671976/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/78fbdd1/ : FAILURE in 9m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a693fb5363afb8219cd55147af007b3b40509114","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-22 12:10:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f492a80c6ece5f57c071f52b791888eab380a80e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 12:20:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/76/671976/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/e5f2942/ : FAILURE in 9m 16s","accounts_in_message":[],"_revision_number":2},{"id":"cd76169644eb241d3bf3d5b5e12b09166ad7f9e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 12:22:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/76/671976/2/check/openstack-tox-linters/2b13c02/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"41fea58dcc93743dfcaf3e313c314e42073d3d4e","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-22 13:00:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-310c27351dd7925f258dd2df2cbda4474e0e7eee : FAILURE in 31m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9b2cf5cc56db257dd8560d80ffb4d5cc0110edb2","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-22 13:14:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1ca5f82b31de70e2983897cf7146b38ffdebe998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 13:26:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/76/671976/3/check/openstack-tox-linters/f7b75c8/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"828d16e3a2434558ca31a85df6843928ad6539d9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 13:26:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/76/671976/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/3c59e57/ : FAILURE in 11m 18s","accounts_in_message":[],"_revision_number":3},{"id":"6f76dfef7fab5016dff22cca491ff4da4c775a81","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-22 13:50:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-9f88c57e959efe2a473816867df59b6e6451eb85 : FAILURE in 34m 50s","accounts_in_message":[],"_revision_number":3},{"id":"8c26b4fa11299927147a8dd97216158b6483832a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-23 06:35:30.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2fbc578035c75d9916b36b4e25d285475014cc67","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 06:46:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/76/671976/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/a37a7fd/ : FAILURE in 10m 37s","accounts_in_message":[],"_revision_number":4},{"id":"21ae2c056ca0f78637f7499b90fce046db511f04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 06:55:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/76/671976/4/check/openstack-tox-linters/72ed021/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":4},{"id":"44d65336b6deeb23e690c1f3320e02eb0ed4207c","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-23 07:19:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-83d82233cada723fe8791343bd290a5ee72f5d49 : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":4},{"id":"5e5aa0f6f54592fb83c75ff2b252a7aec4a48867","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-04 17:54:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6823deb56316e67718086aecae4f38c4d892a39c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 17:57:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/76/671976/4/check/openstack-tox-linters/5f890fd/ : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":4},{"id":"f28da088e25a7e2ea60664febcd7c18076df3f73","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-04 18:05:40.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/76/671976/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/7a2ac8b/ : FAILURE in 10m 21s","accounts_in_message":[],"_revision_number":4},{"id":"cbb23f0e925105d302b86bf12566c2c1fe80a884","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-04 18:22:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-ba0c51923cc9eed47c319e7a17028d6265c7d39e : SUCCESS in 28m 18s","accounts_in_message":[],"_revision_number":4},{"id":"511b70a4ccb02fb6ebcf00b853484ce61e86c08b","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-10 12:53:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1566c0d77e19fd3030b3662824ac494db3b15df9","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-10 12:57:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f7fd3bf976dfa3c42d6d8d6bfdf1f218e3d8345f","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-10 19:04:34.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"5099bbafddb8b2a20fb3254d01ff3d1d0b7bef46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 19:14:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b8d1625b02f64cb7b68c4194a5824af0 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":5},{"id":"d6c1c719afea17aeb15325d6f41392fd796f4d16","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-10 19:19:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/76/671976/5/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/df87e3c/ : FAILURE in 13m 59s","accounts_in_message":[],"_revision_number":5},{"id":"aed03b5b6209ed838f83431b78aa1a0986a1b6b7","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-10 19:27:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-a44a6c25ff384d9cfc0f026d6e7b1bb8ed0f876b : FAILURE in 22m 37s","accounts_in_message":[],"_revision_number":5},{"id":"5172644d7c9390e223834cd206e46de663141438","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-11 18:17:50.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"612c81bde9ec9d5bdf8debacf9ca52685e3c726b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 18:28:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f7146b44934a4d11a0fe3842b2e9e3e4 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":6},{"id":"ca8d98929ea25822a57472db0ae6f87090a7408e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-11 18:32:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/76/671976/6/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/b7fa2e7/ : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":6},{"id":"7b760e422929c92db8297b9ea46ab45f194cd6f0","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 19:11:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-d233b4642d11b0abd9e02b04643e2f8209ed92e9 : FAILURE in 53m 11s","accounts_in_message":[],"_revision_number":6},{"id":"094f01e6c65e989b958fde14dd7369b4efbd766f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-12 12:59:17.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a16181302e14733422ea104bcacdec3d924ded99","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 13:11:52.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/76/671976/6/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/4bb54bd/ : FAILURE in 12m 05s","accounts_in_message":[],"_revision_number":6},{"id":"60127164b86b71116bc537171daf50b2a0111a73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 13:12:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3b281aae7ced4ebf94139adf5e94920d : POST_FAILURE in 10m 33s","accounts_in_message":[],"_revision_number":6},{"id":"0b4bdc60091a21c70b7a4f64d68ecfc6b2a8b1f5","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 19:14:20.000000000","message":"Patch Set 6:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-28ce86be7f5b5f7d1a151da4a415b4148e9b75b3 : FAILURE in 6h 14m 44s","accounts_in_message":[],"_revision_number":6},{"id":"49274ed7539b4d15714c249ddf0ec1153c533af1","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-13 09:02:13.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b35155b4afb48f3084251baa4d0aa39459413f5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 09:11:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8306564ef911491285c78572b6d8cea5 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":6},{"id":"f77a1866d97e9e9a44e95b4b301063ecbda3a519","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-13 09:19:55.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/76/671976/6/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/a56e6ba/ : FAILURE in 17m 12s","accounts_in_message":[],"_revision_number":6},{"id":"cdd28986ac5a7e63f184e9b71d879b31fd348791","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-13 09:29:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-379181ce7877b5327bdc19089628fcad24f0e49a : SUCCESS in 26m 53s","accounts_in_message":[],"_revision_number":6},{"id":"7979b8575eb20cb63e387e5be948b108c14bd8b8","author":{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"},"date":"2020-07-31 13:04:22.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d98d9c0c6d8bb864aaeb19eb7d0529a763deb0b7","author":{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"},"date":"2020-07-31 13:04:50.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"a34f6406576da57880d215ae70a28c21d84e372b","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-07-31 13:17:06.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"571b987ba0a04576749884802945cfcd13876d06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 13:17:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"338c2f133b96cf1af35f97db80846bab1669dce5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 13:40:07.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7f27f77b9dc640988835ed5056b94064 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":6},{"id":"ba8d54ab6074db6cabad988f6bbd05a79bf08949","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 13:40:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d233b4642d11b0abd9e02b04643e2f8209ed92e9","revisions":{"8d698d86b7b35516b4f89031ee194074dee75a47":{"kind":"REWORK","_number":1,"created":"2019-07-22 07:49:36.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/76/671976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/76/671976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/1"}}},"commit":{"parents":[{"commit":"7c7ac3b998e763ea651d2e37149208df7a974018","subject":"Merge \"nova: switch to python3 for build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7c7ac3b998e763ea651d2e37149208df7a974018"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-22 07:49:23.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-22 07:49:23.000000000","tz":120},"subject":"Install the pyc as well for the filelist capturing it","message":"Install the pyc as well for the filelist capturing it\n\nThe change in Ib03f38285c3de377d4c71de4ed7feb1bfa919a35 was actually\nincomplete, it updated the filelist but never actually instaled\nthe additional files, so they were not included.\n\nChange-Id: Ibb3cbfc6c940ba5a436fcb13e8d9200e647acc78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8d698d86b7b35516b4f89031ee194074dee75a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8d698d86b7b35516b4f89031ee194074dee75a47"}]},"branch":"refs/heads/master"},"bb2f8848104b39d1c2d21b18ad6958e08ea76a45":{"kind":"REWORK","_number":2,"created":"2019-07-22 12:10:43.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/76/671976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/76/671976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/2"}}},"commit":{"parents":[{"commit":"7c7ac3b998e763ea651d2e37149208df7a974018","subject":"Merge \"nova: switch to python3 for build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7c7ac3b998e763ea651d2e37149208df7a974018"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-22 07:49:23.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-22 12:10:35.000000000","tz":120},"subject":"Install the pyc as well for the filelist capturing it","message":"Install the pyc as well for the filelist capturing it\n\nThe change in Ib03f38285c3de377d4c71de4ed7feb1bfa919a35 was actually\nincomplete, it updated the filelist but never actually instaled\nthe additional files, so they were not included.\n\nChange-Id: Ibb3cbfc6c940ba5a436fcb13e8d9200e647acc78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/bb2f8848104b39d1c2d21b18ad6958e08ea76a45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/bb2f8848104b39d1c2d21b18ad6958e08ea76a45"}]},"branch":"refs/heads/master"},"dd1edbe2ae5636918d10c463a0174922f625b0ae":{"kind":"REWORK","_number":3,"created":"2019-07-22 13:14:50.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/76/671976/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/76/671976/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/3"}}},"commit":{"parents":[{"commit":"7c7ac3b998e763ea651d2e37149208df7a974018","subject":"Merge \"nova: switch to python3 for build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7c7ac3b998e763ea651d2e37149208df7a974018"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-22 07:49:23.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-22 13:14:41.000000000","tz":120},"subject":"Install the pyc as well for the filelist capturing it","message":"Install the pyc as well for the filelist capturing it\n\nThe change in Ib03f38285c3de377d4c71de4ed7feb1bfa919a35 was actually\nincomplete, it updated the filelist but never actually instaled\nthe additional files, so they were not included.\n\nChange-Id: Ibb3cbfc6c940ba5a436fcb13e8d9200e647acc78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/dd1edbe2ae5636918d10c463a0174922f625b0ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/dd1edbe2ae5636918d10c463a0174922f625b0ae"}]},"branch":"refs/heads/master"},"83d82233cada723fe8791343bd290a5ee72f5d49":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-23 06:35:30.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/76/671976/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/76/671976/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/4"}}},"commit":{"parents":[{"commit":"259690d920b4fac1e7b64cd335d27d16c0e14539","subject":"Merge \"keystone: add new dependency for documentation build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/259690d920b4fac1e7b64cd335d27d16c0e14539"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-22 07:49:23.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-23 06:35:30.000000000","tz":0},"subject":"Install the pyc as well for the filelist capturing it","message":"Install the pyc as well for the filelist capturing it\n\nThe change in Ib03f38285c3de377d4c71de4ed7feb1bfa919a35 was actually\nincomplete, it updated the filelist but never actually instaled\nthe additional files, so they were not included.\n\nChange-Id: Ibb3cbfc6c940ba5a436fcb13e8d9200e647acc78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/83d82233cada723fe8791343bd290a5ee72f5d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/83d82233cada723fe8791343bd290a5ee72f5d49"}]},"branch":"refs/heads/master"},"a44a6c25ff384d9cfc0f026d6e7b1bb8ed0f876b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-09-10 19:04:34.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/76/671976/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/76/671976/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/5"}}},"commit":{"parents":[{"commit":"38a0bc32b74aaba45d50a9367b446d8739784746","subject":"Merge \"neutron: Add missing Requires for conntrack-tools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/38a0bc32b74aaba45d50a9367b446d8739784746"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-22 07:49:23.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-10 19:04:34.000000000","tz":0},"subject":"Install the pyc as well for the filelist capturing it","message":"Install the pyc as well for the filelist capturing it\n\nThe change in Ib03f38285c3de377d4c71de4ed7feb1bfa919a35 was actually\nincomplete, it updated the filelist but never actually instaled\nthe additional files, so they were not included.\n\nChange-Id: Ibb3cbfc6c940ba5a436fcb13e8d9200e647acc78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a44a6c25ff384d9cfc0f026d6e7b1bb8ed0f876b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a44a6c25ff384d9cfc0f026d6e7b1bb8ed0f876b"}]},"branch":"refs/heads/master"},"d233b4642d11b0abd9e02b04643e2f8209ed92e9":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-09-11 18:17:50.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/76/671976/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/76/671976/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/76/671976/6"}}},"commit":{"parents":[{"commit":"a008eb2094836dbeeb7ee67403cbbf2e32f9d07f","subject":"Merge \"Increase unit test timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a008eb2094836dbeeb7ee67403cbbf2e32f9d07f"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-22 07:49:23.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-11 18:17:50.000000000","tz":0},"subject":"Install the pyc as well for the filelist capturing it","message":"Install the pyc as well for the filelist capturing it\n\nThe change in Ib03f38285c3de377d4c71de4ed7feb1bfa919a35 was actually\nincomplete, it updated the filelist but never actually instaled\nthe additional files, so they were not included.\n\nChange-Id: Ibb3cbfc6c940ba5a436fcb13e8d9200e647acc78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d233b4642d11b0abd9e02b04643e2f8209ed92e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d233b4642d11b0abd9e02b04643e2f8209ed92e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
