)]}'
{"id":"openstack%2Frpm-packaging~672951","triplet_id":"openstack%2Frpm-packaging~master~I2ee3730bd8fb78a8c88a032654fb4569081bd677","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"I2ee3730bd8fb78a8c88a032654fb4569081bd677","subject":"Update to the current oslo.* library version releases","status":"MERGED","created":"2019-07-26 11:08:16.000000000","updated":"2019-08-05 19:11:17.000000000","submitted":"2019-08-05 19:11:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672951-1565032277669-d227193f","meta_rev_id":"1f817b5446459225e869373769d434b3185bb569","_number":672951,"virtual_id_number":672951,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-08-02 23:04:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-08-05 19:11:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-02 21:01:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-05 13:29:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-05 19:00:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 21:01:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-02 23:04:32.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-05 13:29:06.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-08-05 19:11:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6bddd6900f11f1f07d5a4430c239345c343a26a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-26 11:08:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10dbff1101f6bf2175bc4d5e10a5a33d7e9e2275","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 11:12:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/51/672951/1/check/openstack-tox-linters/d44b74d/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8090e59329f53251dc6710be68d819192da583f9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-26 11:29:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/51/672951/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/25c277e/ : FAILURE in 20m 31s","accounts_in_message":[],"_revision_number":1},{"id":"622267a4bceb6a3983ca6a2f8377e0a5f3e28dad","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-07-26 11:56:02.000000000","message":"Patch Set 1:\n\nSome patches for os-api-ref seem to be included in the new release.","accounts_in_message":[],"_revision_number":1},{"id":"793f2511289aec7cc479b67e51c14a7738dbd1c7","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-26 12:00:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"33865b650b17ae34c8879ef4bcc3bc387a3df05d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 12:11:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters http://logs.openstack.org/51/672951/2/check/openstack-tox-linters/bc001fd/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5a7e45e0d492782bba2105cf3e8c7c1241d91eb8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-26 13:32:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/51/672951/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/5dc5248/ : TIMED_OUT in 1h 30m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b552fb975ad60968a92f2b9c1f63cf1eb1d25045","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-07-26 15:01:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-88c358cf81479318d42a3a1b10dd73c1f0b99f05 : FAILURE in 2h 59m 48s","accounts_in_message":[],"_revision_number":2},{"id":"a495540c3b3bd685b9715209a86c23ef7c1fa9f9","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-01 11:44:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5e958e96defcd14e8ab8d8beec4291586f5f5bf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 11:56:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/51/672951/3/check/openstack-tox-linters/740b3a0/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"580487583bcac553cb47514fde0e57cdfbfdf6be","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 12:05:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/51/672951/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/0830d3e/ : FAILURE in 20m 18s","accounts_in_message":[],"_revision_number":3},{"id":"4355d6cc2badcda00f7bb1c706aade8212702ee5","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-08-01 12:14:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"37425bad6e4038a8be51e7b049de58f1d3e7256f","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 14:13:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-d1c0b88652a5245b170eb790deb718a23d8151ed : FAILURE in 2h 28m 13s","accounts_in_message":[],"_revision_number":3},{"id":"2fa75d489ac99d4c015e191f02e54df8903683c1","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-01 15:18:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2e00dd928a4cceb0be2eb8a2f0eadac294edf37c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 15:40:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/51/672951/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/60b4af0/ : FAILURE in 20m 55s","accounts_in_message":[],"_revision_number":4},{"id":"4af7818f158aa77b5ac34359fc693f4173b8c60f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 15:42:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/51/672951/4/check/openstack-tox-linters/ddcf933/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":4},{"id":"b8d54a9cd808b7bce8c9d8ee754dc99f88e4d1cd","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 16:53:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-5edf909a7669355209a1a231100ebc0fc49beabd : FAILURE in 1h 34m 40s","accounts_in_message":[],"_revision_number":4},{"id":"52b9d13095c35a911faac4190a0c51a023dd862a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-01 18:15:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c67d77105db7679895fac391190fe159d2824090","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 18:24:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://logs.opendev.org/51/672951/5/check/openstack-tox-linters/a0494ca/ : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":5},{"id":"ccaeaaba3de87cdde453169fc809987ff66a542e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 19:11:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/51/672951/5/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/ff6567d/ : FAILURE in 21m 02s","accounts_in_message":[],"_revision_number":5},{"id":"3b87d48cb60617c69ae0aa9338e195273be222cd","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 20:12:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-7bca2135f710f0c7329c78e316376a84de0a8ad5 : FAILURE in 1h 56m 54s","accounts_in_message":[],"_revision_number":5},{"id":"a317e15ccc27a9668dcd705890b004924df3342a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-02 11:49:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0d147a1c441254d2352b9172df806d56628f1883","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 11:59:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/51/672951/6/check/openstack-tox-linters/0d5aa63/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":6},{"id":"e6d5ba0d71305cc62d605f76b4c6c4c19d3566f4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 12:10:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/51/672951/6/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/a36d813/ : FAILURE in 20m 46s","accounts_in_message":[],"_revision_number":6},{"id":"13f3f287ee6632cf5ff81ca68befda7fc646f33a","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 12:53:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-0d705e14ccc467ef1d75d9af2183dcfa373973cf : FAILURE in 1h 03m 35s","accounts_in_message":[],"_revision_number":6},{"id":"b67c3c85b6a26873fb5538885dcc38d560e0de8d","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-02 14:18:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"eca5489cad2ecfb0dd7fc15cd0120ededf4b5fd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 14:28:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/51/672951/7/check/openstack-tox-linters/e0ba44c/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":7},{"id":"3333123d15a67d111914a844e6aeebce9378c1c9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 14:44:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/51/672951/7/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/e99397a/ : SUCCESS in 25m 41s","accounts_in_message":[],"_revision_number":7},{"id":"76ac0c9b7d59bf53cd81bd7a20d888a1a9f03ca8","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 16:15:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-e85913397ed481c5a708b575f8a2b3c0a70743da : FAILURE in 1h 57m 11s","accounts_in_message":[],"_revision_number":7},{"id":"cf1a635c553b56ecd3d71251444355959ad83917","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-02 20:40:25.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"948bd76f282e52778b769850dd65f75320c35402","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 20:49:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/51/672951/8/check/openstack-tox-linters/bb3a685/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":8},{"id":"53583e2e7e12c967471d8d9c66a3146bc3dd5ec2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 21:01:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/51/672951/8/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/f51e2e4/ : SUCCESS in 19m 42s","accounts_in_message":[],"_revision_number":8},{"id":"32e950d45f92e68370ff57d0ae00491acf1be5d3","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 23:04:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-df977b983ff78e30b1a11a5cd41b12e2fefce36b : FAILURE in 2h 21m 21s","accounts_in_message":[],"_revision_number":8},{"id":"1f2152647f47ebc88a28d28edc0248ac9246a7e2","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-08-05 13:29:06.000000000","message":"Patch Set 8: Code-Review+2\n\nThe SUSE CI errors seem unrelated. The cinder failure should was reported in https://bugs.launchpad.net/cinder/+bug/1838225, and should be fixed soon.","accounts_in_message":[],"_revision_number":8},{"id":"218ca53476d7db19b1d14ed3537fa44de664bc50","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-05 19:00:44.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"91885e7c51e938b97b12d309fe15731fbb3f00c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 19:00:57.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"77ed1cfcb8fe5ed9dedc7e71a796717ba59f6541","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 19:11:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"1f817b5446459225e869373769d434b3185bb569","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 19:11:17.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://logs.opendev.org/51/672951/8/gate/openstack-tox-linters/3ba7d36/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"fe3b9c6f037d111bded3644621f0b108a5ebfcd1","revisions":{"db996ff1169eea2eebf314bc2dfe2ee912b6b29a":{"kind":"REWORK","_number":1,"created":"2019-07-26 11:08:16.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/51/672951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/51/672951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/1"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-26 11:08:03.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-26 11:08:03.000000000","tz":120},"subject":"Latest oslo fun","message":"Latest oslo fun\n\nChange-Id: I2ee3730bd8fb78a8c88a032654fb4569081bd677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/db996ff1169eea2eebf314bc2dfe2ee912b6b29a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/db996ff1169eea2eebf314bc2dfe2ee912b6b29a"}]},"branch":"refs/heads/master"},"88c358cf81479318d42a3a1b10dd73c1f0b99f05":{"kind":"REWORK","_number":2,"created":"2019-07-26 12:00:58.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/51/672951/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/51/672951/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/2"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-26 11:08:03.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-26 12:00:50.000000000","tz":120},"subject":"Latest oslo fun","message":"Latest oslo fun\n\nChange-Id: I2ee3730bd8fb78a8c88a032654fb4569081bd677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/88c358cf81479318d42a3a1b10dd73c1f0b99f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/88c358cf81479318d42a3a1b10dd73c1f0b99f05"}]},"branch":"refs/heads/master"},"d1c0b88652a5245b170eb790deb718a23d8151ed":{"kind":"REWORK","_number":3,"created":"2019-08-01 11:44:52.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/51/672951/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/51/672951/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/3"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-26 11:08:03.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-01 11:44:05.000000000","tz":120},"subject":"Update to the current oslo.* library version releases","message":"Update to the current oslo.* library version releases\n\nos-api-ref: Drop upstreamed patches that are present\nin new release\nos-win: convert to using stestr\n\nChange-Id: I2ee3730bd8fb78a8c88a032654fb4569081bd677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d1c0b88652a5245b170eb790deb718a23d8151ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d1c0b88652a5245b170eb790deb718a23d8151ed"}]},"branch":"refs/heads/master"},"5edf909a7669355209a1a231100ebc0fc49beabd":{"kind":"REWORK","_number":4,"created":"2019-08-01 15:18:42.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/51/672951/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/51/672951/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/4"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-26 11:08:03.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-01 15:18:34.000000000","tz":120},"subject":"Update to the current oslo.* library version releases","message":"Update to the current oslo.* library version releases\n\nos-api-ref: Drop upstreamed patches that are present\nin new release\nos-win: convert to using stestr\n\nChange-Id: I2ee3730bd8fb78a8c88a032654fb4569081bd677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5edf909a7669355209a1a231100ebc0fc49beabd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5edf909a7669355209a1a231100ebc0fc49beabd"}]},"branch":"refs/heads/master"},"7bca2135f710f0c7329c78e316376a84de0a8ad5":{"kind":"REWORK","_number":5,"created":"2019-08-01 18:15:21.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/51/672951/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/51/672951/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/5"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-26 11:08:03.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-01 18:15:13.000000000","tz":120},"subject":"Update to the current oslo.* library version releases","message":"Update to the current oslo.* library version releases\n\nos-api-ref: Drop upstreamed patches that are present\nin new release\nos-win: convert to using stestr\n\nChange-Id: I2ee3730bd8fb78a8c88a032654fb4569081bd677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7bca2135f710f0c7329c78e316376a84de0a8ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7bca2135f710f0c7329c78e316376a84de0a8ad5"}]},"branch":"refs/heads/master"},"6f34dbe3385bcd37280fc13c4730158ee38809b9":{"kind":"REWORK","_number":6,"created":"2019-08-02 11:49:19.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/51/672951/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/51/672951/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/6"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-26 11:08:03.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-02 11:49:02.000000000","tz":120},"subject":"Update to the current oslo.* library version releases","message":"Update to the current oslo.* library version releases\n\nos-api-ref: Drop upstreamed patches that are present\nin new release\nos-win: convert to using stestr\n\nChange-Id: I2ee3730bd8fb78a8c88a032654fb4569081bd677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6f34dbe3385bcd37280fc13c4730158ee38809b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6f34dbe3385bcd37280fc13c4730158ee38809b9"}]},"branch":"refs/heads/master"},"705cca44689d9ffdcc9b782ad400f5232abf802d":{"kind":"REWORK","_number":7,"created":"2019-08-02 14:18:21.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/51/672951/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/51/672951/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/7"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-26 11:08:03.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-02 14:18:05.000000000","tz":120},"subject":"Update to the current oslo.* library version releases","message":"Update to the current oslo.* library version releases\n\nos-api-ref: Drop upstreamed patches that are present\nin new release\nos-win: convert to using stestr\n\nChange-Id: I2ee3730bd8fb78a8c88a032654fb4569081bd677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/705cca44689d9ffdcc9b782ad400f5232abf802d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/705cca44689d9ffdcc9b782ad400f5232abf802d"}]},"branch":"refs/heads/master"},"fe3b9c6f037d111bded3644621f0b108a5ebfcd1":{"kind":"REWORK","_number":8,"created":"2019-08-02 20:40:25.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/51/672951/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/51/672951/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/51/672951/8"}}},"commit":{"parents":[{"commit":"831bb61c23c29413cd98fba11d76c2ff7757873a","subject":"Merge \"Convert python-designateclient to singlespec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/831bb61c23c29413cd98fba11d76c2ff7757873a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-07-26 11:08:03.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-02 20:40:13.000000000","tz":120},"subject":"Update to the current oslo.* library version releases","message":"Update to the current oslo.* library version releases\n\nos-api-ref: Drop upstreamed patches that are present\nin new release\nos-win: convert to using stestr\n\nChange-Id: I2ee3730bd8fb78a8c88a032654fb4569081bd677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/fe3b9c6f037d111bded3644621f0b108a5ebfcd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/fe3b9c6f037d111bded3644621f0b108a5ebfcd1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
