)]}'
{"id":"openstack%2Frpm-packaging~678500","triplet_id":"openstack%2Frpm-packaging~master~I36e6bbc66b432fc83efb396f7268d892315689e1","project":"openstack/rpm-packaging","branch":"master","topic":"neutron-lib","hashtags":[],"change_id":"I36e6bbc66b432fc83efb396f7268d892315689e1","subject":"neuton-lib: Update to 1.29.0","status":"MERGED","created":"2019-08-26 09:33:57.000000000","updated":"2019-08-26 10:56:48.000000000","submitted":"2019-08-26 10:56:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678500-1566817008294-831344a3","meta_rev_id":"a90754a0f9affceb3f0ddb95fc749e3a3d7bceb6","_number":678500,"virtual_id_number":678500,"owner":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-26 10:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-26 10:39:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-26 10:46:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":2,"date":"2019-08-26 10:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":1,"date":"2019-08-26 10:52:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 10:07:35.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-26 10:39:14.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-26 10:46:57.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-08-26 10:52:12.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-08-26 10:56:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08fb2ba3bd35a20a98a39f05d819ff96e7992039","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-08-26 09:33:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94b5a7aef22f7c0411685802fbd5960a1c20c30d","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-26 09:35:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc6880cb56433bebe94393cb71dbd53de3c0dccc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 09:41:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4df2031326a7481a99eab2fa89d19e5c : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b4c0e45481d6890fe6397d281523059574bab463","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 09:44:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/00/678500/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/1dd4ceb/ : FAILURE in 10m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f2c7bbee91d9e0db5916ad7e2c4703d45561442d","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-08-26 09:59:42.000000000","message":"Patch Set 1: Code-Review-1\n\nhttps://opendev.org/openstack/neutron-lib/commit/24aef1398b5b69967696aa8cb6598ab1be37c626 added os-ken as a requirement. We need to include it here as both build and runtime requirement.","accounts_in_message":[],"_revision_number":1},{"id":"e48b137c9fa85c8fd5b5124e586339aabd611474","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-26 10:00:32.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"8f3264a4f02d7bbe62be9ead8accb657e8b107e1","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-26 10:07:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-622b5719b154c2afe476c988a33c9c7eaa5a1a5e : FAILURE in 33m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cdaede7d74ecdfa5301f2aa8d14edafa241bb3cb","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-08-26 10:24:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94328adda629d1f1cecc069407802af094829ab7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 10:32:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/56a79c3d6d664dbc82d9ee3c14f916a1 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a6846978fd508685c0b814ffcfa6e9e95da65296","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 10:39:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/00/678500/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/2be3ed4/ : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2d1095f78da4b144980583c777cf8573ce3fe6ff","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-08-26 10:46:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2841c81ca45455d81acd665d542013c93c6fae81","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-26 10:52:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"90584c793ebe84a59ad6a46f6de2f3c3aa186e1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 10:52:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"215302a2dbc71e9079a00f4f2077614265195d00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 10:56:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5211e2d5160c467cb21f9ac99f601a33 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"99d7fd9766d940834838a0ccd75b9c13bbab6e5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 10:56:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ede4c0c4a2e85ef72c88172853a0fd7b67bfb93c","revisions":{"622b5719b154c2afe476c988a33c9c7eaa5a1a5e":{"kind":"REWORK","_number":1,"created":"2019-08-26 09:33:57.000000000","uploader":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"ref":"refs/changes/00/678500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/00/678500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/00/678500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/00/678500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/00/678500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/00/678500/1"}}},"commit":{"parents":[{"commit":"a61469b6082c0cefe26ba8bce93be5e075db73ab","subject":"Merge \"pymod2pkg: Update to 0.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a61469b6082c0cefe26ba8bce93be5e075db73ab"}]}],"author":{"name":"Ralf Haferkamp","email":"rhafer@suse.de","date":"2019-08-26 09:32:12.000000000","tz":120},"committer":{"name":"Ralf Haferkamp","email":"rhafer@suse.de","date":"2019-08-26 09:32:12.000000000","tz":120},"subject":"neuton-lib: Update to 1.29.0","message":"neuton-lib: Update to 1.29.0\n\nNeutron requires 1.29.0 now\n\nChange-Id: I36e6bbc66b432fc83efb396f7268d892315689e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/622b5719b154c2afe476c988a33c9c7eaa5a1a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/622b5719b154c2afe476c988a33c9c7eaa5a1a5e"}]},"branch":"refs/heads/master"},"ede4c0c4a2e85ef72c88172853a0fd7b67bfb93c":{"kind":"REWORK","_number":2,"created":"2019-08-26 10:24:26.000000000","uploader":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"ref":"refs/changes/00/678500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/00/678500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/00/678500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/00/678500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/00/678500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/00/678500/2"}}},"commit":{"parents":[{"commit":"a61469b6082c0cefe26ba8bce93be5e075db73ab","subject":"Merge \"pymod2pkg: Update to 0.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a61469b6082c0cefe26ba8bce93be5e075db73ab"}]}],"author":{"name":"Ralf Haferkamp","email":"rhafer@suse.de","date":"2019-08-26 09:32:12.000000000","tz":120},"committer":{"name":"Ralf Haferkamp","email":"rhafer@suse.de","date":"2019-08-26 10:23:16.000000000","tz":120},"subject":"neuton-lib: Update to 1.29.0","message":"neuton-lib: Update to 1.29.0\n\nNeutron requires 1.29.0 now. This version also introduced a new\ndependency on \"os-ken\".\n\nChange-Id: I36e6bbc66b432fc83efb396f7268d892315689e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ede4c0c4a2e85ef72c88172853a0fd7b67bfb93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ede4c0c4a2e85ef72c88172853a0fd7b67bfb93c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
