)]}'
{"id":"openstack%2Frpm-packaging~679485","triplet_id":"openstack%2Frpm-packaging~stable%2Fstein~I2abe11fc0abe662f4c4466b3ce15c38a20abd669","project":"openstack/rpm-packaging","branch":"stable/stein","topic":"l2gw-stein","hashtags":[],"change_id":"I2abe11fc0abe662f4c4466b3ce15c38a20abd669","subject":"Packaging for networking-l2gw","status":"MERGED","created":"2019-08-30 17:07:23.000000000","updated":"2019-09-13 05:11:21.000000000","submitted":"2019-09-13 05:11:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679485-1568351481466-e7c50ac1","meta_rev_id":"6034d59d8e8ef34552cdc3cbedc5381ca1108c71","_number":679485,"virtual_id_number":679485,"owner":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-09-09 18:23:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-09-13 05:11:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27532,"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","username":"jaicaa"},{"value":0,"date":"2019-09-10 16:29:32.000000000","_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},{"value":1,"date":"2019-09-09 18:05:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-12 18:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27532,"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","username":"jaicaa"},{"value":0,"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-13 05:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27532,"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","username":"jaicaa"},{"value":0,"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-13 05:05:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27532,"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","username":"jaicaa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 09:05:55.000000000","updated_by":{"_account_id":27532,"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","username":"jaicaa"},"reviewer":{"_account_id":27532,"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","username":"jaicaa"},"state":"REVIEWER"},{"updated":"2019-09-09 18:05:42.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-09 18:23:11.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 18:12:06.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-09-13 05:05:10.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-09-13 05:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2f9c5cc24197ea954a5df31bc80d6fb25813a86","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-08-30 17:07:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ac4048d5f140368e380291039b6c824d6088b9bc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 17:13:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/b031dae/ : FAILURE in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"673368824b6e9cb0a14624cb59209069a73ad1f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 17:16:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/94ca9507b7f04b9aa86b8f43ba031b75 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"89490850d7f07561077c3e902ed28a6cff7c2c81","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 18:17:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-aedd899dfd9051b7c64c8d5cd489a2444c144475 : FAILURE in 1h 09m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c8fb22ec2966f83d9c03482af695df4f098ae3c3","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-08-30 19:14:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e8c8e20aba538990bdf0737a2f850b830a185197","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 19:19:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/6323f71/ : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"66c009d7bf5fc05b87950376686a663ff613d4f1","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 19:20:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-92c6648c2d9e117638d3c1fcc61139b34d3a0411 : FAILURE in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7248348d5e6dc222377e219fe9f0420da87c15e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:22:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/826628ff60ff44b8b86888e197a7ee7b : FAILURE in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"404a5c07276c0dee91b6c29a2e05ff7a73dcc32d","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-08-30 19:40:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"45fa9a46448a4cf97c8128715f7c9f07731fa847","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 19:48:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/93d21e9/ : FAILURE in 6m 38s","accounts_in_message":[],"_revision_number":3},{"id":"4b471373d41f09870f25f2203615e535d696f9a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:48:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0f9afcd4933b4a7e8794278df9ba4387 : FAILURE in 5m 07s","accounts_in_message":[],"_revision_number":3},{"id":"223189fc3357e879ede72059937c7fe82d1ba4f4","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-08-30 20:08:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2b5d320180afb59a94f1ed0ef1acf338525dc822","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 20:15:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/f03b0b5/ : FAILURE in 6m 24s","accounts_in_message":[],"_revision_number":4},{"id":"2bf2c75856096fdc479f2bb63e4e8981f3499afb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:18:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/13a30dc3f6d84386a159ef74b3e124a5 : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":4},{"id":"c1bdb0f7953f756ceed666d6403310b3eb5ce9fd","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-08-30 20:23:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4a2ccb1d5df2b7b0d8be134137d4eee243ab2c3a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 20:31:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/5/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/a37ee8d/ : FAILURE in 6m 51s","accounts_in_message":[],"_revision_number":5},{"id":"9f5217a326aa930b5ba10cf16b38312a9a7bc653","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:33:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/75848e333e494140bb7549f4daa4f632 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":5},{"id":"eac95b5e9c17bd8bb65b710f3c272284d4e0215e","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 20:57:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-0647c76f8851bfd5fab61fb2142a9837e836c5cd : FAILURE in 33m 54s","accounts_in_message":[],"_revision_number":5},{"id":"e1b3cf21980161aa7f92faa69842512d002977eb","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-08-30 21:17:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ed93c23c34cd5742507cccd8eef457e16494459a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 21:24:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/6/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/91926ee/ : FAILURE in 6m 43s","accounts_in_message":[],"_revision_number":6},{"id":"5ad2313ae10263bcc60a750ee0c74fec1dfb84ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 21:27:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd4080ebb9674b19ba65467ea52739b0 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":6},{"id":"bcde3d9aee761144bfe200b6eeac67bbe709167b","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 22:07:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-e746a8fb00620856937596fa0059bc483942b221 : FAILURE in 50m 10s","accounts_in_message":[],"_revision_number":6},{"id":"44d13d751a00cd185ad641aee3652032318fa557","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-03 21:08:21.000000000","message":"Topic changed from SOC-5837 to l2gw-rocky","accounts_in_message":[],"_revision_number":6},{"id":"eff3ada628c3a6d5f43968d970a49bbf60db39f3","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-03 21:08:29.000000000","message":"Topic changed from l2gw-rocky to l2gw-stein","accounts_in_message":[],"_revision_number":6},{"id":"8556fb1cf0cb99a90110e1ef873d194481f4a0b0","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-03 21:13:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"573c0247b070ff543c04d0183bf9108651f7d55d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 21:24:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6de195e57441451cb60f796ae08d04c6 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":7},{"id":"dba2361cd0bbeb29e1cda53b61d4069ac467ab12","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-03 21:31:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/7/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/1291a62/ : FAILURE in 16m 33s","accounts_in_message":[],"_revision_number":7},{"id":"9728949341215e9ea8e921613591f90371b3b788","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-03 21:44:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-d976a0d599e4ab9d500b3dd81df371d159601b52 : FAILURE in 30m 41s","accounts_in_message":[],"_revision_number":7},{"id":"2af545fff4b2663c1ee3f1a139e98c8104b60150","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-03 22:09:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ec1db94b6223358a272b03f07a64df74144457ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-03 22:13:49.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/8/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/f4949a0/ : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":8},{"id":"dde12d496b1b6d18e030ebb83f72579f14107d98","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-03 22:15:17.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-0c8ceecb9a0c8b7eaaef93d87a97dd8e1e1259e5 : FAILURE in 5m 12s","accounts_in_message":[],"_revision_number":8},{"id":"3762e2a05d9e4b1ba42ee24ca3a8c6d783021148","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 22:18:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/79f0021f19944c7f9269fd45ede1db7b : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":8},{"id":"79898a8509bcb8abf8165a68c64b0c7bbac9c1ed","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-03 22:23:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a8a69f1630e606b1835e9710cb4405e9319c5525","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-03 22:30:27.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/9/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/18a86b4/ : FAILURE in 6m 19s","accounts_in_message":[],"_revision_number":9},{"id":"60a493df37b720f0880bac27eadc471dc9df47f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 22:32:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/31cddf24d6f240cfa4cf63fa1dbb0c82 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":9},{"id":"6a7e735523db94eddfcb8890d78ed2fb96dddecf","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-03 23:10:25.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-1830f6f05d7bd2f69ed6b6ba572114077045f03d : FAILURE in 46m 33s","accounts_in_message":[],"_revision_number":9},{"id":"db59d340aa5edaa2c6bf5057288781da51f9e983","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-04 06:55:57.000000000","message":"Patch Set 9: Code-Review-1\n\nyou need to remove the depends-on and backport the sphinx_build macro first.","accounts_in_message":[],"_revision_number":9},{"id":"8757be95c8fbdcd29942e13820ad3cc3a9c63013","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-04 06:56:28.000000000","message":"Patch Set 9:\n\nthis patch needs backport: https://review.opendev.org/#/c/668208/","accounts_in_message":[],"_revision_number":9},{"id":"f4c29476816d74b6ecc297ae0a062627d7b18af6","author":{"_account_id":27532,"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","username":"jaicaa"},"date":"2019-09-04 09:05:55.000000000","message":"Patch Set 9:\n\nAlso I think this needs to use py2_build/py2_install instead of python_builb/python_install","accounts_in_message":[],"_revision_number":9},{"id":"8401f83b4c4251065339afed39e5221f5bef87da","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-04 16:53:36.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f663e44389dc038eb6ba610bd65076698612f8f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-04 16:57:56.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/10/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/36a17d9/ : FAILURE in 3m 00s","accounts_in_message":[],"_revision_number":10},{"id":"b262390ba840110978e5e5100eee145e4735bc0a","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 16:59:34.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-aa8e34012037f86bb0e015243402531f693f33fa : FAILURE in 5m 38s","accounts_in_message":[],"_revision_number":10},{"id":"915875274e7a482b0f464a4e1a6714730aa5709d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 17:04:37.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7ae1afe3b3ca49b3beec67f683b083c3 : FAILURE in 5m 35s","accounts_in_message":[],"_revision_number":10},{"id":"ff8c924e86fe8af26e9e191bbd285ab7285fee70","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-04 17:14:50.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"458d236857f0c955462d08b4bdadfd2bb820f4d3","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-04 17:20:19.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"6ded913e1548f2cf3f435b84210691b794d77116","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-04 17:22:19.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/11/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/25bbaa8/ : FAILURE in 6m 59s","accounts_in_message":[],"_revision_number":11},{"id":"65539710b1efe8938bc692c4c81b21389a99c9ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 17:24:50.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2b0772a7afac4c6ebd4773bdef09c0bf : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":11},{"id":"4175b8c252d1e7574084454e1906efe7b8c0ad12","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 17:28:06.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-a0e9762f9e4575358aedbc770ce161db3a7c5043 : FAILURE in 12m 54s","accounts_in_message":[],"_revision_number":11},{"id":"9de19e55dee0e99b48a2068aa2d407a6070f4793","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-04 21:14:31.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c215a06f0e3b52567406b8ea6e7debb60ad60fab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-04 21:20:59.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/12/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/ce306f2/ : FAILURE in 5m 38s","accounts_in_message":[],"_revision_number":12},{"id":"46da8d5e3dff37d68877c2dcdc7fca0578c16110","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 21:25:46.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a35e6bb6c44d4ae1ade5a6b0a640e657 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":12},{"id":"089c4653b09e339cb82640b4735162d55ceadbec","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 22:40:38.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-8d124f0a5d7275dbec51e10cc1d2853d8ede4ef1 : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":12},{"id":"aab6f7a13b6c9c53f25bac6c9899f3ccfbb9000b","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-04 23:26:22.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"06798b8d0a48a94686bdf6d84698338e709b1a9f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-04 23:32:39.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/13/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/ba026eb/ : FAILURE in 5m 45s","accounts_in_message":[],"_revision_number":13},{"id":"71a56285b84756e8d3cfe7c1ec9a5e3655303a3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 23:35:57.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/23c963a9f84c469faea2d4bbe6f003c6 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":13},{"id":"4ce147c9f70beae9034bfc9c208f1917e6164711","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 23:39:30.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-a6b2aaf98ad4b66321e0854444dc9fd863853771 : FAILURE in 12m 48s","accounts_in_message":[],"_revision_number":13},{"id":"9f81440932738a6449f7a7b6618bc25bab8073d3","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-05 00:35:16.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9ab7c61a2f8162d3e68f85aaae338d03736a8205","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-05 00:42:26.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/14/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/388577f/ : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":14},{"id":"fcbbd28b07eb44153e95d031d45447fec4615511","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 00:48:34.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/be89e239ff83433a84b1db4176e715c8 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":14},{"id":"60ed37a5f0b0ee0740a9ec6ded1e2a2f008d4bcc","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-05 00:54:02.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-a2b5d6271b23dbb35f4e708a36a41f63025cc618 : SUCCESS in 18m 25s","accounts_in_message":[],"_revision_number":14},{"id":"ee5ea1ca41fd2446a3bef3b14eb3220f6190b2aa","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-05 13:50:42.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1c515f685d32896711fc430247787ea5e01793f8","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-05 14:25:32.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"4bbdd0a79daafbc4a026f52f084a752b07b2b190","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-05 14:26:04.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8128c75c39267fb9f51abe08a830e8cf5fd6d428","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-05 14:32:59.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/15/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/ffcfeaa/ : FAILURE in 6m 52s","accounts_in_message":[],"_revision_number":15},{"id":"56b5d02f9d4c3798aeb5ec8575e22e4f312998cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 14:37:00.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/43d2d83a0f084475adac3e4e0c19c7ff : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":15},{"id":"3aea018e8ec7fd21fe3f959a683794befd39d602","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-05 15:02:06.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-12c32810b60402136fa4dcfe919c8029cc022062 : FAILURE in 32m 56s","accounts_in_message":[],"_revision_number":15},{"id":"a6c3ad0b86cf3d894e35673cbabdc27b1c1d4d4c","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-05 15:12:56.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"427601ff73fcd9bde89d4191da5ee812a2990c03","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-05 15:44:42.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"854e84058197af6a80f10059487865446ef3ce40","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-05 15:58:06.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/16/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/aee08d4/ : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":16},{"id":"78e7ee05dcbd3c45fa48356f8ca706c3cc2e3974","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-05 16:13:18.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-4b80988b1e67d9d64445af5bc3466217f8a7081c : SUCCESS in 28m 16s","accounts_in_message":[],"_revision_number":16},{"id":"82c32cdb07c0e52c070dedb89f5a2ad592eb9b86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 16:56:52.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7fa44941cbfc41ccaf8e8c90c39b0a0e : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":16},{"id":"a7ca5f206820e5737444908d8747f55880646e88","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-05 17:30:34.000000000","message":"Patch Set 16: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 0c1efca4a05a35eaf019ad973916b751ab3461c6","accounts_in_message":[],"_revision_number":16},{"id":"95daf9ab8c55d80cc4e640c1efeb91d780eb9790","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-06 10:49:18.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"e6e1470419fdf67e7cd12426b0ccf089043a36fe","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-06 18:06:21.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"e51240da6909c6a1a686ab7ab86fa17fc7f3443e","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-09 08:12:08.000000000","message":"Patch Set 16: -Code-Review","accounts_in_message":[],"_revision_number":16},{"id":"ba1a17462f8d4e630c26fa0ff8b7a3d31ec9a0c3","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-09 17:55:40.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"07fe432c95ae96557dc27e45004461bfdd423ad8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-09 18:05:42.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-centos http://logs.rdoproject.org/85/679485/17/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-centos/45aa289/ : SUCCESS in 9m 08s","accounts_in_message":[],"_revision_number":17},{"id":"194e6d0b360146b74589ab1063a8acd46d30c229","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 18:07:49.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6c717052e4564555a2a25b87d60051dd : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":17},{"id":"203bd1865c08cacb1464a3a1aa3b16847c051213","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-09-09 18:23:11.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Stein https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Stein-8b7904ca76d0a77c47174e8a83c79bc999cb8bbe : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":17},{"id":"d027e918b6eb31ed3ce9462c7306c5b7b9f0fb81","author":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"date":"2019-09-10 16:29:32.000000000","message":"Patch Set 17:\n\nPS17 corrected the py2pkg statements in the spec file templates.","accounts_in_message":[],"_revision_number":17},{"id":"fb645fd00a531a6849e4e35ad01657615897e8f8","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-12 18:12:06.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"6e5f24270e659fc6b944732d6c6495a64d55cad2","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-13 05:05:10.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"d8b6aba601fa938518a0983349e94cc6806bf806","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 05:05:25.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"4ec880ff48cb699dd03f319811b465eb792c6bff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 05:11:21.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5587cbc618949a68dbb7d20732464d0 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":17},{"id":"6034d59d8e8ef34552cdc3cbedc5381ca1108c71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 05:11:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"8b7904ca76d0a77c47174e8a83c79bc999cb8bbe","revisions":{"aedd899dfd9051b7c64c8d5cd489a2444c144475":{"kind":"REWORK","_number":1,"created":"2019-08-30 17:07:23.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/1"}}},"commit":{"parents":[{"commit":"839cd6a39fbf196903c96ff9038c98a49397dc46","subject":"Merge \"watcher: add systemd units for watcher services\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/839cd6a39fbf196903c96ff9038c98a49397dc46"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"stephen-ma","email":"sma@suse.com","date":"2019-08-30 17:07:23.000000000","tz":0},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\nDepends-On: Iba6bebccd933b134cb04dcc8f197feac84b272bd\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/aedd899dfd9051b7c64c8d5cd489a2444c144475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/aedd899dfd9051b7c64c8d5cd489a2444c144475"}]},"branch":"refs/heads/stable/stein"},"92c6648c2d9e117638d3c1fcc61139b34d3a0411":{"kind":"REWORK","_number":2,"created":"2019-08-30 19:14:52.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/2"}}},"commit":{"parents":[{"commit":"839cd6a39fbf196903c96ff9038c98a49397dc46","subject":"Merge \"watcher: add systemd units for watcher services\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/839cd6a39fbf196903c96ff9038c98a49397dc46"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-08-30 19:13:06.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\nDepends-On: Iba6bebccd933b134cb04dcc8f197feac84b272bd\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/92c6648c2d9e117638d3c1fcc61139b34d3a0411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/92c6648c2d9e117638d3c1fcc61139b34d3a0411"}]},"branch":"refs/heads/stable/stein"},"0c238c7948257d94355e196d41ecbcb075e69aff":{"kind":"REWORK","_number":3,"created":"2019-08-30 19:40:56.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/3"}}},"commit":{"parents":[{"commit":"839cd6a39fbf196903c96ff9038c98a49397dc46","subject":"Merge \"watcher: add systemd units for watcher services\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/839cd6a39fbf196903c96ff9038c98a49397dc46"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-08-30 19:40:33.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\nDepends-On: Iba6bebccd933b134cb04dcc8f197feac84b272bd\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0c238c7948257d94355e196d41ecbcb075e69aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0c238c7948257d94355e196d41ecbcb075e69aff"}]},"branch":"refs/heads/stable/stein"},"ed6df7157e25562004713fba6bdbb1a25069b296":{"kind":"REWORK","_number":4,"created":"2019-08-30 20:08:12.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/4"}}},"commit":{"parents":[{"commit":"839cd6a39fbf196903c96ff9038c98a49397dc46","subject":"Merge \"watcher: add systemd units for watcher services\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/839cd6a39fbf196903c96ff9038c98a49397dc46"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-08-30 20:07:57.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\nDepends-On: Iba6bebccd933b134cb04dcc8f197feac84b272bd\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ed6df7157e25562004713fba6bdbb1a25069b296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ed6df7157e25562004713fba6bdbb1a25069b296"}]},"branch":"refs/heads/stable/stein"},"0647c76f8851bfd5fab61fb2142a9837e836c5cd":{"kind":"REWORK","_number":5,"created":"2019-08-30 20:23:30.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/5"}}},"commit":{"parents":[{"commit":"839cd6a39fbf196903c96ff9038c98a49397dc46","subject":"Merge \"watcher: add systemd units for watcher services\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/839cd6a39fbf196903c96ff9038c98a49397dc46"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-08-30 20:23:18.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\nDepends-On: Iba6bebccd933b134cb04dcc8f197feac84b272bd\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0647c76f8851bfd5fab61fb2142a9837e836c5cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0647c76f8851bfd5fab61fb2142a9837e836c5cd"}]},"branch":"refs/heads/stable/stein"},"e746a8fb00620856937596fa0059bc483942b221":{"kind":"REWORK","_number":6,"created":"2019-08-30 21:17:18.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/6"}}},"commit":{"parents":[{"commit":"839cd6a39fbf196903c96ff9038c98a49397dc46","subject":"Merge \"watcher: add systemd units for watcher services\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/839cd6a39fbf196903c96ff9038c98a49397dc46"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-08-30 21:17:03.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\nDepends-On: Iba6bebccd933b134cb04dcc8f197feac84b272bd\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/e746a8fb00620856937596fa0059bc483942b221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/e746a8fb00620856937596fa0059bc483942b221"}]},"branch":"refs/heads/stable/stein"},"d976a0d599e4ab9d500b3dd81df371d159601b52":{"kind":"REWORK","_number":7,"created":"2019-09-03 21:13:42.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/7"}}},"commit":{"parents":[{"commit":"839cd6a39fbf196903c96ff9038c98a49397dc46","subject":"Merge \"watcher: add systemd units for watcher services\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/839cd6a39fbf196903c96ff9038c98a49397dc46"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-03 21:13:24.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\nDepends-On: Iba6bebccd933b134cb04dcc8f197feac84b272bd\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d976a0d599e4ab9d500b3dd81df371d159601b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d976a0d599e4ab9d500b3dd81df371d159601b52"}]},"branch":"refs/heads/stable/stein"},"0c8ceecb9a0c8b7eaaef93d87a97dd8e1e1259e5":{"kind":"REWORK","_number":8,"created":"2019-09-03 22:09:45.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/8"}}},"commit":{"parents":[{"commit":"839cd6a39fbf196903c96ff9038c98a49397dc46","subject":"Merge \"watcher: add systemd units for watcher services\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/839cd6a39fbf196903c96ff9038c98a49397dc46"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-03 22:09:30.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\nDepends-On: Iba6bebccd933b134cb04dcc8f197feac84b272bd\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0c8ceecb9a0c8b7eaaef93d87a97dd8e1e1259e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0c8ceecb9a0c8b7eaaef93d87a97dd8e1e1259e5"}]},"branch":"refs/heads/stable/stein"},"1830f6f05d7bd2f69ed6b6ba572114077045f03d":{"kind":"REWORK","_number":9,"created":"2019-09-03 22:23:31.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/9"}}},"commit":{"parents":[{"commit":"839cd6a39fbf196903c96ff9038c98a49397dc46","subject":"Merge \"watcher: add systemd units for watcher services\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/839cd6a39fbf196903c96ff9038c98a49397dc46"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-03 22:23:23.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\nDepends-On: Iba6bebccd933b134cb04dcc8f197feac84b272bd\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1830f6f05d7bd2f69ed6b6ba572114077045f03d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1830f6f05d7bd2f69ed6b6ba572114077045f03d"}]},"branch":"refs/heads/stable/stein"},"b05b7bad4e90c5019721dcc2754137da19638b8d":{"kind":"REWORK","_number":10,"created":"2019-09-04 16:53:36.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/10"}}},"commit":{"parents":[{"commit":"839cd6a39fbf196903c96ff9038c98a49397dc46","subject":"Merge \"watcher: add systemd units for watcher services\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/839cd6a39fbf196903c96ff9038c98a49397dc46"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-04 16:36:05.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b05b7bad4e90c5019721dcc2754137da19638b8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b05b7bad4e90c5019721dcc2754137da19638b8d"}]},"branch":"refs/heads/stable/stein"},"a0e9762f9e4575358aedbc770ce161db3a7c5043":{"kind":"REWORK","_number":11,"created":"2019-09-04 17:14:50.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/11"}}},"commit":{"parents":[{"commit":"6985aa7d7923719cb556570ac745dd2a74e6bc25","subject":"Add sphinx_build macro for better redhat/suse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6985aa7d7923719cb556570ac745dd2a74e6bc25"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-04 17:08:03.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a0e9762f9e4575358aedbc770ce161db3a7c5043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a0e9762f9e4575358aedbc770ce161db3a7c5043"}]},"branch":"refs/heads/stable/stein"},"8d124f0a5d7275dbec51e10cc1d2853d8ede4ef1":{"kind":"REWORK","_number":12,"created":"2019-09-04 21:14:31.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/12"}}},"commit":{"parents":[{"commit":"6985aa7d7923719cb556570ac745dd2a74e6bc25","subject":"Add sphinx_build macro for better redhat/suse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6985aa7d7923719cb556570ac745dd2a74e6bc25"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-04 21:13:59.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8d124f0a5d7275dbec51e10cc1d2853d8ede4ef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8d124f0a5d7275dbec51e10cc1d2853d8ede4ef1"}]},"branch":"refs/heads/stable/stein"},"a6b2aaf98ad4b66321e0854444dc9fd863853771":{"kind":"REWORK","_number":13,"created":"2019-09-04 23:26:22.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/13"}}},"commit":{"parents":[{"commit":"6985aa7d7923719cb556570ac745dd2a74e6bc25","subject":"Add sphinx_build macro for better redhat/suse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6985aa7d7923719cb556570ac745dd2a74e6bc25"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-04 23:25:57.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a6b2aaf98ad4b66321e0854444dc9fd863853771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a6b2aaf98ad4b66321e0854444dc9fd863853771"}]},"branch":"refs/heads/stable/stein"},"a2b5d6271b23dbb35f4e708a36a41f63025cc618":{"kind":"REWORK","_number":14,"created":"2019-09-05 00:35:16.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/14"}}},"commit":{"parents":[{"commit":"6985aa7d7923719cb556570ac745dd2a74e6bc25","subject":"Add sphinx_build macro for better redhat/suse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6985aa7d7923719cb556570ac745dd2a74e6bc25"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-05 00:35:08.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a2b5d6271b23dbb35f4e708a36a41f63025cc618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a2b5d6271b23dbb35f4e708a36a41f63025cc618"}]},"branch":"refs/heads/stable/stein"},"12c32810b60402136fa4dcfe919c8029cc022062":{"kind":"REWORK","_number":15,"created":"2019-09-05 14:25:32.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/15"}}},"commit":{"parents":[{"commit":"6985aa7d7923719cb556570ac745dd2a74e6bc25","subject":"Add sphinx_build macro for better redhat/suse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6985aa7d7923719cb556570ac745dd2a74e6bc25"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-05 14:24:51.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/12c32810b60402136fa4dcfe919c8029cc022062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/12c32810b60402136fa4dcfe919c8029cc022062"}]},"branch":"refs/heads/stable/stein"},"4b80988b1e67d9d64445af5bc3466217f8a7081c":{"kind":"REWORK","_number":16,"created":"2019-09-05 15:44:42.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/16"}}},"commit":{"parents":[{"commit":"6985aa7d7923719cb556570ac745dd2a74e6bc25","subject":"Add sphinx_build macro for better redhat/suse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6985aa7d7923719cb556570ac745dd2a74e6bc25"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-05 15:44:27.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/4b80988b1e67d9d64445af5bc3466217f8a7081c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/4b80988b1e67d9d64445af5bc3466217f8a7081c"}]},"branch":"refs/heads/stable/stein"},"8b7904ca76d0a77c47174e8a83c79bc999cb8bbe":{"kind":"REWORK","_number":17,"created":"2019-09-09 17:55:40.000000000","uploader":{"_account_id":6876,"name":"stephen-ma","email":"stephenbma43@gmail.com","username":"stephen-ma"},"ref":"refs/changes/85/679485/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/85/679485/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/85/679485/17"}}},"commit":{"parents":[{"commit":"6985aa7d7923719cb556570ac745dd2a74e6bc25","subject":"Add sphinx_build macro for better redhat/suse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6985aa7d7923719cb556570ac745dd2a74e6bc25"}]}],"author":{"name":"Jaime Caamaño Ruiz","email":"jcaamano@suse.com","date":"2019-07-03 12:38:08.000000000","tz":120},"committer":{"name":"Stephen Ma","email":"sma@suse.com","date":"2019-09-09 17:55:14.000000000","tz":-420},"subject":"Packaging for networking-l2gw","message":"Packaging for networking-l2gw\n\nChange-Id: I2abe11fc0abe662f4c4466b3ce15c38a20abd669\n(cherry picked from commit 3be5f56fbd9858113fd80fa177ed0b6175e951a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8b7904ca76d0a77c47174e8a83c79bc999cb8bbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8b7904ca76d0a77c47174e8a83c79bc999cb8bbe"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
