)]}'
{"id":"openstack%2Frpm-packaging~686088","triplet_id":"openstack%2Frpm-packaging~master~I5128b3c842daed03d9ddf641609c70ae6c30c269","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"I5128b3c842daed03d9ddf641609c70ae6c30c269","subject":"Add initial spec for zaqar","status":"MERGED","created":"2019-10-02 06:44:06.000000000","updated":"2019-10-03 15:02:51.000000000","submitted":"2019-10-03 15:02:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686088-1570114971466-39d839bb","meta_rev_id":"b37d987c77603d281f5ef168861729fd73a1dd47","_number":686088,"virtual_id_number":686088,"owner":{"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-10-02 18:23:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 15:02:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-02 17:32:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-03 14:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-03 08:15:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-03 14:57:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 17:32:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 18:23:45.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-03 08:15:04.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-10-03 14:57:45.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-10-03 15:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90bb31a33617ef1940b7907e4ef89c561643b537","author":{"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},"date":"2019-10-02 06:44:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6af1356228053e31c1653d0daeb30f4768fb6a26","author":{"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},"date":"2019-10-02 06:49:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa55ca6597e39fe0e733fde1f5a8611d14d90dd1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-02 06:57:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/88/686088/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/1e1e372/ : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"abf9392cb082bac23115044ed4f74191b64d8394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 06:59:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8db0a8291cdc463f95a54737056e98de : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"0f94c512605dbfa6e98f209765b889c8637fd26d","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-02 07:36:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-fae29d70124e231a4eee58c51baf7a55ddc447b8 : SUCCESS in 46m 41s","accounts_in_message":[],"_revision_number":2},{"id":"2e2268f37847196e5eb2f2f5bea84e9faad0981d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-02 15:15:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ba589cab2cbe7a22b73df55c010334be5f18d493","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-02 17:18:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nJust one nit, other than that it looks ok to me.","accounts_in_message":[],"_revision_number":2},{"id":"00a3afd7d2dbe959e3ad4bd7cae5ea4a7bb84849","author":{"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},"date":"2019-10-02 17:26:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e50ccd72b4c7b81943bfd17a3b28358535417e1a","author":{"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},"date":"2019-10-02 17:28:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5abd9614223b67ef84edf6c99023375fc64207fd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-02 17:32:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/88/686088/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/08e198b/ : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":3},{"id":"1f679f05bc26664447156b7ec84b18cf65cd84b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 18:09:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a569095e8a314c98a0cde0cf238766a0 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"085c9997d2cfb2d3e3e69b777b361287be14935e","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-02 18:23:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-30843b9008d76e81fccc391d545dfe49322b40b5 : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":3},{"id":"5f2c203ab66e4f7932c4be0c3b865ea3f15ca7f8","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-03 08:15:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0f12ddee2f550398369cd9e08f6eae5ade54bc49","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-03 14:57:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d16c8b2664845b99977779894d493d6fcfc82b16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 14:57:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"14a355d419e550b54a72dea13c62cbdfed29445e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:02:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7b1495d475924a58afe2af49a3cc5d84 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"deebe8944755317de2c3c4b64e0e5108abd20811","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:02:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"30843b9008d76e81fccc391d545dfe49322b40b5","revisions":{"76be8bc3229948e4f352a2c801db5af482884ae6":{"kind":"REWORK","_number":1,"created":"2019-10-02 06:44:06.000000000","uploader":{"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},"ref":"refs/changes/88/686088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/686088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/1"}}},"commit":{"parents":[{"commit":"54dfbae016c760fcd14c453e9173a7e5d87c7d6c","subject":"Merge \"Initial spec for swift\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/54dfbae016c760fcd14c453e9173a7e5d87c7d6c"}]}],"author":{"name":"matthew-fuller","email":"mfuller@suse.com","date":"2019-10-02 06:43:33.000000000","tz":-420},"committer":{"name":"matthew-fuller","email":"mfuller@suse.com","date":"2019-10-02 06:43:33.000000000","tz":-420},"subject":"Add initial spec for zaqar","message":"Add initial spec for zaqar\n\nChange-Id: I5128b3c842daed03d9ddf641609c70ae6c30c269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/76be8bc3229948e4f352a2c801db5af482884ae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/76be8bc3229948e4f352a2c801db5af482884ae6"}]},"branch":"refs/heads/master"},"fae29d70124e231a4eee58c51baf7a55ddc447b8":{"kind":"REWORK","_number":2,"created":"2019-10-02 06:49:40.000000000","uploader":{"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},"ref":"refs/changes/88/686088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/686088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/2"}}},"commit":{"parents":[{"commit":"54dfbae016c760fcd14c453e9173a7e5d87c7d6c","subject":"Merge \"Initial spec for swift\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/54dfbae016c760fcd14c453e9173a7e5d87c7d6c"}]}],"author":{"name":"matthew-fuller","email":"mfuller@suse.com","date":"2019-10-02 06:43:33.000000000","tz":-420},"committer":{"name":"matthew-fuller","email":"mfuller@suse.com","date":"2019-10-02 06:49:19.000000000","tz":-420},"subject":"Add initial spec for zaqar","message":"Add initial spec for zaqar\n\nChange-Id: I5128b3c842daed03d9ddf641609c70ae6c30c269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/fae29d70124e231a4eee58c51baf7a55ddc447b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/fae29d70124e231a4eee58c51baf7a55ddc447b8"}]},"branch":"refs/heads/master"},"30843b9008d76e81fccc391d545dfe49322b40b5":{"kind":"REWORK","_number":3,"created":"2019-10-02 17:26:10.000000000","uploader":{"_account_id":28777,"name":"Matthew Fuller","email":"matfuller@microsoft.com","username":"mfuller"},"ref":"refs/changes/88/686088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/686088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/686088/3"}}},"commit":{"parents":[{"commit":"54dfbae016c760fcd14c453e9173a7e5d87c7d6c","subject":"Merge \"Initial spec for swift\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/54dfbae016c760fcd14c453e9173a7e5d87c7d6c"}]}],"author":{"name":"matthew-fuller","email":"mfuller@suse.com","date":"2019-10-02 06:43:33.000000000","tz":-420},"committer":{"name":"matthew-fuller","email":"mfuller@suse.com","date":"2019-10-02 17:25:49.000000000","tz":-420},"subject":"Add initial spec for zaqar","message":"Add initial spec for zaqar\n\nChange-Id: I5128b3c842daed03d9ddf641609c70ae6c30c269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/30843b9008d76e81fccc391d545dfe49322b40b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/30843b9008d76e81fccc391d545dfe49322b40b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
