)]}'
{"id":"openstack%2Frpm-packaging~687054","triplet_id":"openstack%2Frpm-packaging~master~I72c00b9cbb3b5b6d56eb134224c7f183f043c980","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"I72c00b9cbb3b5b6d56eb134224c7f183f043c980","subject":"cyborg: Switch to python3","status":"MERGED","created":"2019-10-07 13:50:59.000000000","updated":"2019-10-08 04:32:52.000000000","submitted":"2019-10-08 04:32:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687054-1570509172413-14372de2","meta_rev_id":"2202c0ac082427d044921aee7628d04b557915e0","_number":687054,"virtual_id_number":687054,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-10-08 04:29:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 04:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-08 04:17:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-08 04:27:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-08 04:27:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 15:09:32.000000000","updated_by":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"reviewer":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"state":"REVIEWER"},{"updated":"2019-10-07 15:18:33.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-10-08 04:17:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-08 04:29:13.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-08 04:32:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d7e9e8538c8c0ff16246e4fa9e447103d93341b","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-10-07 13:50:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3af3a73c7b83946e8023a70d504ea04a464351fc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-07 13:59:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/54/687054/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/b16ec8e/ : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"cb6cff734a53d278c533faea122bf0e63656251c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 14:00:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/573080bfd53e4a3d8fb3d5797d336462 : FAILURE in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"609e18990bedccd7dfb3ae118c42f066c299153c","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-07 14:45:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-7f0713f307b39cd345bf1ba8d58868c155631cde : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":1},{"id":"092d2ed0b43ba3f0e60376b729c80a8e61128fd8","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-10-07 15:08:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"190889bc82922dd06d43779df7b899accd96bad1","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-10-07 15:09:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ad548a6fa704b9f4851ee8786441ca2d82421bfd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-07 15:17:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/54/687054/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/e2d5b21/ : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":2},{"id":"b38349bf47169546759d22a9bcf1f655e930cb3f","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-07 15:18:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"faa2654cf294a8b509782896747fb685b1a845e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 15:23:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/85df2add3dd74f1494400ef5a28317be : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"90738cc9cfc98f9dad6cefc49f3b6b1d1026745d","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-07 15:24:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-106b9f20822529bf279ad76cd886a32cd7de311b : SUCCESS in 15m 27s","accounts_in_message":[],"_revision_number":2},{"id":"8b591afffbbd0113d98536d0e445cdb6752731fe","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-10-08 04:08:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d65590208876325d186858b16e2c2c2660d5690","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-10-08 04:09:07.000000000","message":"Patch Set 2:\n\nFixed only the linting problem","accounts_in_message":[],"_revision_number":2},{"id":"2c92d3f3a5fbae09aefa47049d1261a15cdb9919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 04:13:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1f6c7087640b4fc5b71e226363b14070 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":3},{"id":"d7581f6ad40e45625785e6478aa9d276b7895ef4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-08 04:17:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/54/687054/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/8fec305/ : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":3},{"id":"27982ec0c5f25a4eb76cf9b8d26c5849110d6cec","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-10-08 04:27:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nMerging given that it had already 2x +2 and I only fixed linting ...","accounts_in_message":[],"_revision_number":3},{"id":"dcb10613a65deda65c0d45aaaf4298ea15515fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 04:27:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1c0f18ac26c536401578c9b1ce60de17509e9639","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-08 04:29:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-4d90d6e077c402e5430d0300515994bac2a5f016 : SUCCESS in 20m 01s","accounts_in_message":[],"_revision_number":3},{"id":"30a08c0471f324aea99e7db9ee66a8e7544f644c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 04:32:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/751153305a354b93b5c499a0aa845a60 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2202c0ac082427d044921aee7628d04b557915e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 04:32:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ed91395910498c6a79d874e8f45047ca45258f7a","revisions":{"7f0713f307b39cd345bf1ba8d58868c155631cde":{"kind":"REWORK","_number":1,"created":"2019-10-07 13:50:59.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/54/687054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/54/687054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/1"}}},"commit":{"parents":[{"commit":"3432a157367db1951f94a4904e49d97e99abe470","subject":"Merge \"python-dracclient: Update to 3.1.1 and switch to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3432a157367db1951f94a4904e49d97e99abe470"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-10-07 13:50:37.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-10-07 13:50:37.000000000","tz":120},"subject":"cyborg: Switch to python3","message":"cyborg: Switch to python3\n\nChange-Id: I72c00b9cbb3b5b6d56eb134224c7f183f043c980\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7f0713f307b39cd345bf1ba8d58868c155631cde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7f0713f307b39cd345bf1ba8d58868c155631cde"}]},"branch":"refs/heads/master"},"106b9f20822529bf279ad76cd886a32cd7de311b":{"kind":"REWORK","_number":2,"created":"2019-10-07 15:08:44.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/54/687054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/54/687054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/2"}}},"commit":{"parents":[{"commit":"61e60899423715628786c35556c0228603ec9c82","subject":"Merge \"freezer-tempest-plugin: Switch to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/61e60899423715628786c35556c0228603ec9c82"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-10-07 13:50:37.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-10-07 15:08:31.000000000","tz":120},"subject":"cyborg: Switch to python3","message":"cyborg: Switch to python3\n\nChange-Id: I72c00b9cbb3b5b6d56eb134224c7f183f043c980\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/106b9f20822529bf279ad76cd886a32cd7de311b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/106b9f20822529bf279ad76cd886a32cd7de311b"}]},"branch":"refs/heads/master"},"ed91395910498c6a79d874e8f45047ca45258f7a":{"kind":"REWORK","_number":3,"created":"2019-10-08 04:08:53.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/54/687054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/54/687054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/54/687054/3"}}},"commit":{"parents":[{"commit":"61e60899423715628786c35556c0228603ec9c82","subject":"Merge \"freezer-tempest-plugin: Switch to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/61e60899423715628786c35556c0228603ec9c82"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-10-07 13:50:37.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-10-08 04:08:45.000000000","tz":120},"subject":"cyborg: Switch to python3","message":"cyborg: Switch to python3\n\nChange-Id: I72c00b9cbb3b5b6d56eb134224c7f183f043c980\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ed91395910498c6a79d874e8f45047ca45258f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ed91395910498c6a79d874e8f45047ca45258f7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
