)]}'
{"id":"openstack%2Frpm-packaging~688143","triplet_id":"openstack%2Frpm-packaging~master~I08d30c3a70971d54a618a548ae0cc66736dfc3c2","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"I08d30c3a70971d54a618a548ae0cc66736dfc3c2","subject":"heat-agents: convert to python 3","status":"ABANDONED","created":"2019-10-11 15:11:38.000000000","updated":"2019-10-21 07:34:42.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9f39e1c5fb09978798e2ce8c0e1fdea5b6044b6d","_number":688143,"virtual_id_number":688143,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"disliked":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-21 07:34:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-21 07:34:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 13:07:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-15 13:10:34.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-15 13:47:33.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-17 12:17:01.000000000","updated_by":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"reviewer":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"state":"REVIEWER"},{"updated":"2019-10-18 09:00:50.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"}],"messages":[{"id":"7f38e9fd86a7cbacc1e5c226e4c3d23313f13a2f","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-11 15:11:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16c5184aa227008f7ef8f699320cedff33a7d223","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 15:18:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/43/688143/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/40358f8/ : FAILURE in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"37b84a97a76d6f398973f9b9910e4c2d2882eed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 15:21:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97e9522cce6548c7a55d32872f7ca543 : FAILURE in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"90fb2ac0c28cf7c97e34232cb72aba8ee5a66961","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-11 15:25:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-e3ecca5ffc5b9f930a86a60ae74aa149d965554d : FAILURE in 13m 27s","accounts_in_message":[],"_revision_number":1},{"id":"09233146febbc11410a1adbaa2159c87b72c0381","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-11 15:33:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1ae6a733bc7173a0d51cf64d0e5f672207f008ec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 15:40:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/43/688143/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/f652dd0/ : FAILURE in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"06228288f35f0ae92b73431637be91037ae6daaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 15:42:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/04c7042261ad495d915fe6358aa5bf62 : FAILURE in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"93dd62fb96bc558cf654879a8e830f0a6c2a6a46","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-11 16:41:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-5b4bcf933bb4ba75faa0e6d66d8083b312272e79 : FAILURE in 1h 07m 39s","accounts_in_message":[],"_revision_number":2},{"id":"874303933f3f9b649b6a6c9756cbe8c9406a4dea","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-15 10:46:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"699d711dd09cbf1f297d2e50bdd640ff84320079","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 10:50:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/43/688143/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/b41cf5e/ : FAILURE in 2m 18s","accounts_in_message":[],"_revision_number":3},{"id":"be3f3ab68dcde594b9d8e073dbf483bdff98b0d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 10:51:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/485f134efc4e425180211d583d0cbb40 : FAILURE in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"710e5d5a3ec6fa9711e2a8dbe3aa3b81ab0119cc","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-15 11:57:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-1d5e20aa60560f7640bb64af2963d4544c904c11 : FAILURE in 1h 10m 23s","accounts_in_message":[],"_revision_number":3},{"id":"e07d485303cfa060f0c4d27baa00aa5e764c06d8","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-15 12:55:17.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"85a10181778cdd52f0f4fa1cdfac3e9a786b27cd","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-15 12:59:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-f20d7d9b444099d63108cb463bed174bd321cc60 : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":4},{"id":"e1666c7384e62092f685d58b29a8e73359cd2343","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 13:00:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/43/688143/4/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/6891ca9/ : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":4},{"id":"8af66d222ced71a0efd5a199368461b4b3c546a0","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-15 13:03:31.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0fc01d189892763dc98c41ceb83112a2e6cc5042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 13:07:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bfa21f5d76d546c0a1dd385ea372d0d7 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":5},{"id":"76850e0dd35187a4dc742cda0fcc3eff3d8b0057","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 13:10:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/43/688143/5/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/b2c6b61/ : FAILURE in 5m 21s","accounts_in_message":[],"_revision_number":5},{"id":"f6b62a7353c4f691f7556d5c730aaa6b885feea4","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-15 13:47:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-83b9e2502dbcc3c3d356e65b1710f1a240d135f6 : FAILURE in 43m 38s","accounts_in_message":[],"_revision_number":5},{"id":"79c5b363abc61bd6c3d38850fc689c50dae470e9","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-10-17 12:17:01.000000000","message":"Patch Set 5: Code-Review+1\n\nThe SUSE ci failure seems unrelated (issue with the too new cmd2 release)\n\nThe fedora failure seems to be because this needs https://review.opendev.org/#/c/688142/","accounts_in_message":[],"_revision_number":5},{"id":"099a22fa75302839ffc2282dc201d78e3f42427f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-10-18 09:00:50.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nI\u0027ll merge this into the other review, otherwise this will never pass","accounts_in_message":[],"_revision_number":5},{"id":"3a0da12452b83379de0066eea405fc1e61bf16fb","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-10-21 07:34:33.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"9f39e1c5fb09978798e2ce8c0e1fdea5b6044b6d","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-10-21 07:34:42.000000000","message":"Abandoned\n\nalready merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1f55ff21b0db0170df375d41f2835abf72444879","revisions":{"e3ecca5ffc5b9f930a86a60ae74aa149d965554d":{"kind":"REWORK","_number":1,"created":"2019-10-11 15:11:38.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/43/688143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/43/688143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/1"}}},"commit":{"parents":[{"commit":"959c8a57e4f8051671d2e93680c5696b4e7f3a90","subject":"os-refresh-config: convert to python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/959c8a57e4f8051671d2e93680c5696b4e7f3a90"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 15:11:21.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 15:11:21.000000000","tz":120},"subject":"heat-agents: convert to python 3","message":"heat-agents: convert to python 3\n\nChange-Id: I08d30c3a70971d54a618a548ae0cc66736dfc3c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/e3ecca5ffc5b9f930a86a60ae74aa149d965554d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/e3ecca5ffc5b9f930a86a60ae74aa149d965554d"}]},"branch":"refs/heads/master"},"5b4bcf933bb4ba75faa0e6d66d8083b312272e79":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-11 15:33:37.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/43/688143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/43/688143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/2"}}},"commit":{"parents":[{"commit":"b303e0101412a6b8bd6aa5b2e24ab3375fc68c24","subject":"os-refresh-config: convert to python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b303e0101412a6b8bd6aa5b2e24ab3375fc68c24"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 15:11:21.000000000","tz":120},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-10-11 15:33:37.000000000","tz":0},"subject":"heat-agents: convert to python 3","message":"heat-agents: convert to python 3\n\nChange-Id: I08d30c3a70971d54a618a548ae0cc66736dfc3c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5b4bcf933bb4ba75faa0e6d66d8083b312272e79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5b4bcf933bb4ba75faa0e6d66d8083b312272e79"}]},"branch":"refs/heads/master"},"1aa0047ffc6dfa5e9473bff61bb35bb3162a9bec":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-15 10:46:38.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/43/688143/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/43/688143/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/3"}}},"commit":{"parents":[{"commit":"0f4f4d45cac7b670c785fba5ef25b025ead22df1","subject":"os-refresh-config: convert to python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0f4f4d45cac7b670c785fba5ef25b025ead22df1"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 15:11:21.000000000","tz":120},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-10-15 10:46:38.000000000","tz":0},"subject":"heat-agents: convert to python 3","message":"heat-agents: convert to python 3\n\nChange-Id: I08d30c3a70971d54a618a548ae0cc66736dfc3c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1aa0047ffc6dfa5e9473bff61bb35bb3162a9bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1aa0047ffc6dfa5e9473bff61bb35bb3162a9bec"}]},"branch":"refs/heads/master"},"51938586cb4b05fe1789648fa9ba32e4e0841ba5":{"kind":"REWORK","_number":4,"created":"2019-10-15 12:55:17.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/43/688143/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/43/688143/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/4"}}},"commit":{"parents":[{"commit":"0f4f4d45cac7b670c785fba5ef25b025ead22df1","subject":"os-refresh-config: convert to python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0f4f4d45cac7b670c785fba5ef25b025ead22df1"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 15:11:21.000000000","tz":120},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-10-15 12:55:15.000000000","tz":0},"subject":"heat-agents: convert to python 3","message":"heat-agents: convert to python 3\n\nChange-Id: I08d30c3a70971d54a618a548ae0cc66736dfc3c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/51938586cb4b05fe1789648fa9ba32e4e0841ba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/51938586cb4b05fe1789648fa9ba32e4e0841ba5"}]},"branch":"refs/heads/master"},"0db01c800544fc080d2145f559aabe05523bf072":{"kind":"REWORK","_number":5,"created":"2019-10-15 13:03:31.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/43/688143/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/43/688143/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/5"}}},"commit":{"parents":[{"commit":"0f4f4d45cac7b670c785fba5ef25b025ead22df1","subject":"os-refresh-config: convert to python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0f4f4d45cac7b670c785fba5ef25b025ead22df1"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 15:11:21.000000000","tz":120},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-10-15 13:03:29.000000000","tz":0},"subject":"heat-agents: convert to python 3","message":"heat-agents: convert to python 3\n\nChange-Id: I08d30c3a70971d54a618a548ae0cc66736dfc3c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0db01c800544fc080d2145f559aabe05523bf072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0db01c800544fc080d2145f559aabe05523bf072"}]},"branch":"refs/heads/master"},"1f55ff21b0db0170df375d41f2835abf72444879":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-10-21 07:34:33.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/43/688143/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/43/688143/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/43/688143/6"}}},"commit":{"parents":[{"commit":"1c38e229e14e71cd360ddc0b87b001384e937b42","subject":"Merge \"os-refresh-config: convert to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1c38e229e14e71cd360ddc0b87b001384e937b42"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 15:11:21.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-10-21 07:34:33.000000000","tz":0},"subject":"heat-agents: convert to python 3","message":"heat-agents: convert to python 3\n\nChange-Id: I08d30c3a70971d54a618a548ae0cc66736dfc3c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1f55ff21b0db0170df375d41f2835abf72444879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1f55ff21b0db0170df375d41f2835abf72444879"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
