)]}'
{"id":"openstack%2Frpm-packaging~688165","triplet_id":"openstack%2Frpm-packaging~master~I612bd05c7393b1ae487e89bd74b61ef7225fd086","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"I612bd05c7393b1ae487e89bd74b61ef7225fd086","subject":"openstack-placement: make the template more generic","status":"MERGED","created":"2019-10-11 16:21:35.000000000","updated":"2019-10-14 10:32:51.000000000","submitted":"2019-10-14 07:04:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688165-1571036671883-25c6a380","meta_rev_id":"f6a7dccdb29d0fbe865c41db501f52ce42298219","_number":688165,"virtual_id_number":688165,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-10-11 18:08:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 07:04:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-11 17:09:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-11 17:04:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":2,"date":"2019-10-14 06:55:07.000000000","_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":1,"date":"2019-10-14 06:55:07.000000000","_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 17:04:08.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-10-11 17:09:50.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-11 18:08:21.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-14 06:55:07.000000000","updated_by":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"reviewer":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"state":"REVIEWER"},{"updated":"2019-10-14 07:04:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f092a67c89f443ecf8147111341b078f188326d","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-11 16:21:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"579a4ffcb2b420c5953a7a210cf04363810c7e0f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 16:29:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/65/688165/1/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/f3c2103/ : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d5ab4953b7b503cc5dda0c20398fb0496b1fcd73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 16:31:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/38f61ff7b87c4cba8562435824194165 : FAILURE in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2b9887621f8e117f0b31a793f799f2383db61f23","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-11 16:52:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3225694ffbb1ecb910072a4361d84a7794523c35","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 16:59:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/65/688165/2/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/36d9387/ : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"246b017a7449bcaaf75169d2057ce669fc6eafed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 17:02:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb17ef65a34845638fd7b125e99c7c08 : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1dc9288e7ce8f763dfdb3a9db203e3e4de51ac6c","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-11 17:02:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"30d401d801c842d747cf1f806d89903dc1a1cdc9","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-10-11 17:04:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b2f825928ef7d877e0b74333cd5abe5d58e15b37","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 17:09:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-DLRN-rpmbuild-rpm-packaging-fedora http://logs.rdoproject.org/65/688165/3/openstack-check-rdo/legacy-DLRN-rpmbuild-rpm-packaging-fedora/51a8181/ : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":3},{"id":"85080b6ee841727489f1ee3806634f7fe9bd3c3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 17:12:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b6faedd0003f47cdbc73fee420d43cd9 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"ad5be970f574844884ae748f2c6e21541d9aa1ca","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-10-11 18:08:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-4c593802397168ba9e3001f7dc2f06be45a65bf8 : FAILURE in 1h 05m 43s","accounts_in_message":[],"_revision_number":3},{"id":"e45f12edc7c791a273ed361d54465ae7459a38af","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-10-14 06:55:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b3651b625f5d37f92ecffc395865e4efd610053e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 06:55:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea40a99e8b64f59958ab93118b9e4ae4e5035769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 07:04:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb566b89a6d94cd38bc25d9ba5fcfc5f : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"d7aeabdec831530aed5fbb38f2bf8ed0518389b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 07:04:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f6a7dccdb29d0fbe865c41db501f52ce42298219","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-10-14 10:32:51.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit f9b931af537dec64912a372536116cc49914c417","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4c593802397168ba9e3001f7dc2f06be45a65bf8","revisions":{"a516f1cd32418f1b292bcdeebec86f7cbe35b047":{"kind":"REWORK","_number":1,"created":"2019-10-11 16:21:35.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/65/688165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/65/688165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/1"}}},"commit":{"parents":[{"commit":"5c65ef1906778f5a25e3cfed3cd2ace1cf3a85d6","subject":"openstacksdk: Fix suse_version detection for mkisofs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5c65ef1906778f5a25e3cfed3cd2ace1cf3a85d6"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 16:20:24.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 16:21:28.000000000","tz":120},"subject":"openstack-placement: fix scripts","message":"openstack-placement: fix scripts\n\nThe service macros included with the spec do not work fine on RDO, since\nthey are not included for Fedora/CentOS. Moving to the scripts used\nfor both SUSE and RDO.\n\nChange-Id: I612bd05c7393b1ae487e89bd74b61ef7225fd086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a516f1cd32418f1b292bcdeebec86f7cbe35b047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a516f1cd32418f1b292bcdeebec86f7cbe35b047"}]},"branch":"refs/heads/master"},"6163af59062153cfe7e2735e34337674682e3063":{"kind":"REWORK","_number":2,"created":"2019-10-11 16:52:04.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/65/688165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/65/688165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/2"}}},"commit":{"parents":[{"commit":"5c65ef1906778f5a25e3cfed3cd2ace1cf3a85d6","subject":"openstacksdk: Fix suse_version detection for mkisofs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5c65ef1906778f5a25e3cfed3cd2ace1cf3a85d6"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 16:20:24.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 16:51:18.000000000","tz":120},"subject":"openstack-placement: make the template more generic","message":"openstack-placement: make the template more generic\n\nThe service macros included with the spec do not work fine on RDO, since\nthey are not included for Fedora/CentOS. Moving to the scripts used\nfor both SUSE and RDO.\n\nAlso:\n- Use apache macros instead of hardcoded directories\n- Do not create files under /srv/www on RDO\n\nChange-Id: I612bd05c7393b1ae487e89bd74b61ef7225fd086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6163af59062153cfe7e2735e34337674682e3063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6163af59062153cfe7e2735e34337674682e3063"}]},"branch":"refs/heads/master"},"4c593802397168ba9e3001f7dc2f06be45a65bf8":{"kind":"REWORK","_number":3,"created":"2019-10-11 17:02:18.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/65/688165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/65/688165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/65/688165/3"}}},"commit":{"parents":[{"commit":"5c65ef1906778f5a25e3cfed3cd2ace1cf3a85d6","subject":"openstacksdk: Fix suse_version detection for mkisofs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5c65ef1906778f5a25e3cfed3cd2ace1cf3a85d6"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 16:20:24.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2019-10-11 17:02:12.000000000","tz":120},"subject":"openstack-placement: make the template more generic","message":"openstack-placement: make the template more generic\n\nThe service macros included with the spec do not work fine on RDO, since\nthey are not included for Fedora/CentOS. Moving to the scripts used\nfor both SUSE and RDO.\n\nAlso:\n- Use apache macros instead of hardcoded directories\n- Do not create files under /srv/www on RDO\n\nChange-Id: I612bd05c7393b1ae487e89bd74b61ef7225fd086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/4c593802397168ba9e3001f7dc2f06be45a65bf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/4c593802397168ba9e3001f7dc2f06be45a65bf8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
