)]}'
{"id":"openstack%2Frpm-packaging~694981","triplet_id":"openstack%2Frpm-packaging~stable%2Focata~Iae37884287bdfff4774dc7e3060bf663773a7fa1","project":"openstack/rpm-packaging","branch":"stable/ocata","topic":"fix-monasca-tempest","hashtags":[],"change_id":"Iae37884287bdfff4774dc7e3060bf663773a7fa1","subject":"Monasca: fix Tempest tests","status":"MERGED","created":"2019-11-19 10:46:38.000000000","updated":"2019-11-20 13:11:02.000000000","submitted":"2019-11-20 13:11:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694981-1574255462464-ded4c5e8","meta_rev_id":"760948ed7d7f0287d7f4de2ac0fa214e952cb8b3","_number":694981,"virtual_id_number":694981,"owner":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-11-19 17:24:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-20 13:11:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-19 17:08:31.000000000","_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-20 13:02:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},{"value":2,"date":"2019-11-19 18:16:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-20 13:02:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 17:24:03.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-19 18:16:25.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-11-20 13:02:00.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-11-20 13:11:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1ea6e0b12e21c1a6db67a8873c61541449d07dc","author":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"date":"2019-11-19 10:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43714994633e65d0bf6de0c67e00142460fed3a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 10:55:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fbe0a531299f4fa0b0e49f78e6c6547a : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3e6ea965c1c5bcd1d0f36a8712ecc25cd23a3513","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-11-19 11:04:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Ocata https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Ocata-4a0d15ee2e8818a204bd14f6784289a6051c1ba6 : FAILURE in 17m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e92832dca14b6707ed0b43fe5add5aaa5c48440c","author":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"date":"2019-11-19 16:36:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dbd05c668ab45eacc3749447300e67376b14956e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 16:47:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/af9af4026d4042128f551707010fd3d2 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"5059860dc4627de31ace30de4ed8f369b37f9286","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-11-19 16:52:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Ocata https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Ocata-ee4c5fe8ecccc778d17e70273e363f92866525d9 : FAILURE in 15m 36s","accounts_in_message":[],"_revision_number":2},{"id":"3122f71712b4613ac7ffbb0667fa1e19d3890771","author":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"date":"2019-11-19 17:08:31.000000000","message":"Patch Set 2:\n\nrecheck-suse","accounts_in_message":[],"_revision_number":2},{"id":"9c3df1d4a84e5a13f56ec2d4256955f0c4574742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 17:16:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4f51f4bb3cf448c9bf6f34f4121fc4b8 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"6b7560a0a0f062c797485ac903b7743d09327906","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2019-11-19 17:24:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles12-Ocata https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles12-Ocata-349a3e4283bf9df5e44fcdaf5fb6e2d56ca29a65 : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":2},{"id":"7c6aae8d07268c79d8cb98b36bd10f36ef198deb","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-11-19 18:16:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fe9c9dcc4c33f22b0e52d61e241cdc6ff38e126c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-11-20 13:02:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"abfe191a790b9f87f0e9792156497af9975309e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 13:02:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a931b0281e8b380d33d1934a7d7ba1bc51842d72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 13:11:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/578ed3406a414656804ee2891fb0d5f0 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"8bc9f5fe7bdfae253f995979fb62ea54115404cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 13:11:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"963adf8959f8d53cf40f49f5517f426f338241ea","revisions":{"4a0d15ee2e8818a204bd14f6784289a6051c1ba6":{"kind":"REWORK","_number":1,"created":"2019-11-19 10:46:38.000000000","uploader":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"ref":"refs/changes/81/694981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/81/694981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/81/694981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/81/694981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/81/694981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/81/694981/1"}}},"commit":{"parents":[{"commit":"32c10ef064225f8d1d8b8d91239ace9931a178f2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/32c10ef064225f8d1d8b8d91239ace9931a178f2"}]}],"author":{"name":"Johannes Grassler","email":"johannes.grassler@suse.com","date":"2019-11-19 09:59:45.000000000","tz":60},"committer":{"name":"Johannes Grassler","email":"johannes.grassler@suse.com","date":"2019-11-19 10:35:21.000000000","tz":60},"subject":"Monasca: fix Tempest tests","message":"Monasca: fix Tempest tests\n\nThis commit introduces two patches for fixing the Monasca\ntempest tests. Both are unlikely to make it into upstream\nstable/ocata at this point so submitting them here:\n\n* monasca-api/fix-metric-name-offset.patch: this one is\n  specific to the unique constellation on SUSE OpenStack Cloud\n  7 where Tempest is from stable/newton but Monasca is from\n  Ocata and thus wrapped in a conditional. See the patch file\n  itself for a detailed explanation of the issue.\n\n* monasca-log-api/fix-tempest-region.patch: this is a backport\n  of its moral equivalent in monasca-tempest-plugin:\n    https://review.opendev.org/#/c/679225\n\nApart from that, this commit removes\n\n  0001-Fix-InfluxDB-repository-list_dimension_values-to-sup.patch\n\nwhich has been merged upstream.\n\nChange-Id: Iae37884287bdfff4774dc7e3060bf663773a7fa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/4a0d15ee2e8818a204bd14f6784289a6051c1ba6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/4a0d15ee2e8818a204bd14f6784289a6051c1ba6"}]},"branch":"refs/heads/stable/ocata"},"963adf8959f8d53cf40f49f5517f426f338241ea":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-19 16:36:55.000000000","uploader":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"ref":"refs/changes/81/694981/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/81/694981/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/81/694981/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/81/694981/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/81/694981/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/81/694981/2"}}},"commit":{"parents":[{"commit":"32c10ef064225f8d1d8b8d91239ace9931a178f2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/32c10ef064225f8d1d8b8d91239ace9931a178f2"}]}],"author":{"name":"Johannes Grassler","email":"johannes.grassler@suse.com","date":"2019-11-19 09:59:45.000000000","tz":60},"committer":{"name":"Johannes Grassler","email":"johannes.grassler@suse.com","date":"2019-11-19 16:35:48.000000000","tz":60},"subject":"Monasca: fix Tempest tests","message":"Monasca: fix Tempest tests\n\nThis commit introduces two patches for fixing the Monasca\ntempest tests. Both are unlikely to make it into upstream\nstable/ocata at this point so submitting them here:\n\n* monasca-api/fix-metric-name-offset.patch: this one is\n  specific to the unique constellation on SUSE OpenStack Cloud\n  7 where Tempest is from stable/newton but Monasca is from\n  Ocata and thus wrapped in a conditional. See the patch file\n  itself for a detailed explanation of the issue.\n\n* monasca-log-api/fix-tempest-region.patch: this is a backport\n  of its moral equivalent in monasca-tempest-plugin:\n    https://review.opendev.org/#/c/679225\n\nApart from that, this commit removes\n\n  0001-Fix-InfluxDB-repository-list_dimension_values-to-sup.patch\n\nwhich has been merged upstream.\n\nDepends-On: Ibb8d218804e348e48bdd9203839a43b099a7585e\nChange-Id: Iae37884287bdfff4774dc7e3060bf663773a7fa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/963adf8959f8d53cf40f49f5517f426f338241ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/963adf8959f8d53cf40f49f5517f426f338241ea"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
