)]}'
{"id":"openstack%2Frpm-packaging~703599","triplet_id":"openstack%2Frpm-packaging~master~I951b42f902e946c485506d2f87f91e2e72038e80","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"I951b42f902e946c485506d2f87f91e2e72038e80","subject":"Fix django-compressor references","status":"MERGED","created":"2020-01-21 12:33:48.000000000","updated":"2020-03-03 08:50:09.000000000","submitted":"2020-02-28 05:25:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703599-1582867512561-28c1a561","meta_rev_id":"47f757e733230e111d93d02ebe32596ad8a04a22","_number":703599,"virtual_id_number":703599,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2020-02-27 19:36:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 05:25:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-27 19:40:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-27 19:21:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-27 19:21:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":2,"date":"2020-02-27 19:21:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":1,"date":"2020-02-28 05:19:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 09:04:39.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-02-20 07:34:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-27 19:36:13.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-27 19:40:34.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-28 05:19:21.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2020-02-28 05:25:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8648fdb29ca6cdefe272c071eff428ec6127d62","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-01-21 12:33:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2848680da12bea8de16863e2a583a07c30b3ef98","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-21 12:43:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/6103dfc60a8e45f2a89a379ccefedeb1 : FAILURE in 9m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4e79907878d28d74abe6811ad84a322fe91b3ba2","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-01-21 12:48:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-ce10ce66c6571a951e08fbef64af69968b2c5a82 : FAILURE in 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"53f2fa01bdd9652b8723e12b4fe9584c3e8ace95","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-21 12:53:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c01ee0d46e61fbf913b54af99f0e882fc3144b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 13:01:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dff24f22a1e34c28a6625f9f66dabada : FAILURE in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"adec920aa97ea407ec6352e3aa67e9d18c6dacea","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-01-21 13:26:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6f515ff00be204f4fe77025ab23e40f0abe385f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-21 13:34:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/0387fda041824fada2dafb69b79ca57a : FAILURE in 7m 29s","accounts_in_message":[],"_revision_number":2},{"id":"e5c4627fe500accb13b9ea2133901cb4f979bac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 13:35:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f24ad0fd73f34e7da9e6334d5b998f90 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"6ac02d04ebc07e31b5103f0915fbcca23b605270","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-01-21 13:41:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-4cd1a3dd66cc6b5bc0f07451895a8d191a1628f0 : FAILURE in 14m 43s","accounts_in_message":[],"_revision_number":2},{"id":"69ba9fbbe24a1acff3b1296e13b969f23fdb5420","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-01-21 13:49:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d52e827e52247fccfd809d9ed1558b103e1f172","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-01-22 17:29:52.000000000","message":"Patch Set 2:\n\nI\u0027m doing some more research on this. Apparently, we need django \u003e\u003d 2.2, since Horizon dropped support for earlier versions in https://opendev.org/openstack/horizon/commit/a2a3e8b6c35c001c5a34ffef6c0a57225014bca9 .","accounts_in_message":[],"_revision_number":2},{"id":"be67aa0b44320da5ace22359e407200da6675232","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-02-07 18:48:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f679962f6680ed48975f893cc72c01cc6a2cbc6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 18:57:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8080848bce514b8cb838d49acaddc5a6 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":3},{"id":"7af5fab7f8305732a328bc0e3a03d20a9f8c2ccc","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-07 19:06:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-51cd3d2ecd6b3a9578ba6ef2e911f84a9b995fc1 : FAILURE in 16m 55s","accounts_in_message":[],"_revision_number":3},{"id":"c1dd785c5313742ad5273aaeed93545b7fa04883","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-07 21:01:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/55eab5bcdcbb4522b441ca45ab553057 : FAILURE in 10m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1a98276a5bfaf3ae67e3f4e8966c288d1e126d8a","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-10 09:04:39.000000000","message":"Patch Set 3: Code-Review+2\n\nWe\u0027ll fix the RDO CI with the migration to CentOS 8","accounts_in_message":[],"_revision_number":3},{"id":"2896d600a1e06db64cc52d7aa2af8e772a199a6b","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-02-14 07:10:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4e77cf97c2f4a13fa9e18c1af2aae2439329efd9","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-02-17 04:58:57.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ec35ca36eca1aa30850fddb973f6fb6393b61213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 05:04:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5934dbbdf0ec44ea89923e7374b20fa1 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":4},{"id":"296e21234df60c523752baf3356db0774b0ae679","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-17 05:07:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/669f3242b75b4ca683b68b810678461f : FAILURE in 8m 12s","accounts_in_message":[],"_revision_number":4},{"id":"8eff50f4aa96cbb738a018d6e594f57f4534948c","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-17 05:32:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-695dcf37d4b7a3bb6275dfbe4b59e57e38e5d3e7 : FAILURE in 33m 04s","accounts_in_message":[],"_revision_number":4},{"id":"1174abc14088fe3ec47ce84d1bced8bece56854a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-20 07:34:36.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ed768a4fa209bf2d4bfb442dfeabc66588636671","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-02-27 19:21:27.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"1e943192c1f9c12002c14f8eb1ad57fdd092f874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 19:29:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/222e708043c74ce384d257ce161fe70b : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":5},{"id":"b102cccd419f1dc2dd3ccee2b560e7c17f483cc0","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-27 19:36:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-26bce62d1227600ca6ece452e8631023aa6806b2 : FAILURE in 14m 25s","accounts_in_message":[],"_revision_number":5},{"id":"c70094f0900aa27c71f1441f7b543972983f3538","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-27 19:40:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/89b8e7362710499d9ced90dd05dbc195 : SUCCESS in 17m 41s","accounts_in_message":[],"_revision_number":5},{"id":"aa3602929400344ce7683e3a7f00c96d1af386fd","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-02-28 05:19:21.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"edeb85173a994e125b3dc6992876c831a521c096","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 05:19:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"095f5983a6e6c1d5aa86968ba990958883c29d11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 05:25:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"c6e5cb0253478e6703e15076efb00e687b272c6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 05:25:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97c5b13fcf6647769906e19e0f5f4442 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":5},{"id":"47f757e733230e111d93d02ebe32596ad8a04a22","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-03 08:50:09.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 2c337635341f40964b48baf36fc735a5772e2fae","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"26bce62d1227600ca6ece452e8631023aa6806b2","revisions":{"ce10ce66c6571a951e08fbef64af69968b2c5a82":{"kind":"REWORK","_number":1,"created":"2020-01-21 12:33:48.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/99/703599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/99/703599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/1"}}},"commit":{"parents":[{"commit":"70aa556c29708ddecf42a1c6bc3edb502fb2d84c","subject":"Merge \"Convert remaining XStatic packages to py3, remove angular-ui-router\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/70aa556c29708ddecf42a1c6bc3edb502fb2d84c"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-01-21 12:33:21.000000000","tz":60},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-01-21 12:33:21.000000000","tz":60},"subject":"Fix django-compressor references","message":"Fix django-compressor references\n\nThe pypi project was renamed from django_compressor to\ndjango-compressor, adjust\n\nChange-Id: I951b42f902e946c485506d2f87f91e2e72038e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ce10ce66c6571a951e08fbef64af69968b2c5a82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ce10ce66c6571a951e08fbef64af69968b2c5a82"}]},"branch":"refs/heads/master"},"4cd1a3dd66cc6b5bc0f07451895a8d191a1628f0":{"kind":"REWORK","_number":2,"created":"2020-01-21 13:26:08.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/99/703599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/99/703599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/2"}}},"commit":{"parents":[{"commit":"70aa556c29708ddecf42a1c6bc3edb502fb2d84c","subject":"Merge \"Convert remaining XStatic packages to py3, remove angular-ui-router\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/70aa556c29708ddecf42a1c6bc3edb502fb2d84c"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-01-21 12:33:21.000000000","tz":60},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-01-21 13:25:58.000000000","tz":60},"subject":"Fix django-compressor references","message":"Fix django-compressor references\n\nThe pypi project was renamed from django_compressor to\ndjango-compressor, adjust\n\nChange-Id: I951b42f902e946c485506d2f87f91e2e72038e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/4cd1a3dd66cc6b5bc0f07451895a8d191a1628f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/4cd1a3dd66cc6b5bc0f07451895a8d191a1628f0"}]},"branch":"refs/heads/master"},"51cd3d2ecd6b3a9578ba6ef2e911f84a9b995fc1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-07 18:48:48.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/99/703599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/99/703599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/3"}}},"commit":{"parents":[{"commit":"1281e952e7f7531e1fc374d66084a2c58f763e79","subject":"Merge \"oslo.db: convert to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1281e952e7f7531e1fc374d66084a2c58f763e79"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-01-21 12:33:21.000000000","tz":60},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-02-07 18:48:48.000000000","tz":0},"subject":"Fix django-compressor references","message":"Fix django-compressor references\n\nThe pypi project was renamed from django_compressor to\ndjango-compressor, adjust\n\nChange-Id: I951b42f902e946c485506d2f87f91e2e72038e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/51cd3d2ecd6b3a9578ba6ef2e911f84a9b995fc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/51cd3d2ecd6b3a9578ba6ef2e911f84a9b995fc1"}]},"branch":"refs/heads/master"},"695dcf37d4b7a3bb6275dfbe4b59e57e38e5d3e7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-02-17 04:58:57.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/99/703599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/99/703599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/4"}}},"commit":{"parents":[{"commit":"6fd19d3a84134e79e4cc9c456c1056c1b6266ec9","subject":"Merge \"fwaas-dashboard: skip selenium integration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6fd19d3a84134e79e4cc9c456c1056c1b6266ec9"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-01-21 12:33:21.000000000","tz":60},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2020-02-17 04:58:57.000000000","tz":0},"subject":"Fix django-compressor references","message":"Fix django-compressor references\n\nThe pypi project was renamed from django_compressor to\ndjango-compressor, adjust\n\nChange-Id: I951b42f902e946c485506d2f87f91e2e72038e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/695dcf37d4b7a3bb6275dfbe4b59e57e38e5d3e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/695dcf37d4b7a3bb6275dfbe4b59e57e38e5d3e7"}]},"branch":"refs/heads/master"},"26bce62d1227600ca6ece452e8631023aa6806b2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-02-27 19:21:27.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/99/703599/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/99/703599/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/99/703599/5"}}},"commit":{"parents":[{"commit":"2d7d7d2a6b6e59af0c9231c37ea3d5beb0189163","subject":"Merge \"cliff: update to 2.18.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/2d7d7d2a6b6e59af0c9231c37ea3d5beb0189163"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-01-21 12:33:21.000000000","tz":60},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2020-02-27 19:21:27.000000000","tz":0},"subject":"Fix django-compressor references","message":"Fix django-compressor references\n\nThe pypi project was renamed from django_compressor to\ndjango-compressor, adjust\n\nChange-Id: I951b42f902e946c485506d2f87f91e2e72038e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/26bce62d1227600ca6ece452e8631023aa6806b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/26bce62d1227600ca6ece452e8631023aa6806b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
