)]}'
{"id":"openstack%2Frpm-packaging~707088","triplet_id":"openstack%2Frpm-packaging~master~Ie28a92823b0a5340789e19860e5d01938807c9fa","project":"openstack/rpm-packaging","branch":"master","topic":"remove-python2","hashtags":[],"change_id":"Ie28a92823b0a5340789e19860e5d01938807c9fa","subject":"debtcollector: convert to python3","status":"MERGED","created":"2020-02-11 09:55:57.000000000","updated":"2020-03-13 15:53:45.000000000","submitted":"2020-02-11 15:35:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707088-1581435335491-4f5f8f67","meta_rev_id":"657d27799f983ba6e9aea403b6aae722660395b9","_number":707088,"virtual_id_number":707088,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-11 15:35:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-11 15:02:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-11 14:56:27.000000000","_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":2,"date":"2020-02-11 15:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":1,"date":"2020-02-11 15:28:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 11:58:14.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-02-11 15:02:23.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-11 15:28:57.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2020-02-11 15:35:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e782eb0a0299c3597812caa40435ee31784b7022","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-11 09:55:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4019aea949c46cef09f6e9da589499a0c03a32dc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-11 09:58:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/e5e68c55c2d84550bacbd11abba99e48 : POST_FAILURE in 1m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6387e1eaed731d042a0ae4129d356d1320338a4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 10:04:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/beecec9ff0244bf8ab7fd0914a0ff181 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"783af7db46d4b1e1e63247957e928d3fce53d4b5","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-11 10:15:16.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"83f8198cf11452766ee7ea86df17e2bc4fc61c9d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-11 10:35:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/32ddccd606d04f749c348e7a457b9f0c : POST_FAILURE in 1m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1689024c99a43fdfa93c88fc66189e8892239b24","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-11 10:52:20.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"772b9562e63ac77d23e227c0037e2b319c4b16e7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-11 10:54:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/e83e433232c34a318edc8c2295ca8529 : POST_FAILURE in 1m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b8fdce16c88c37f7b6bb9a9c3c647956e3db758f","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-11 11:29:39.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0122372670108156c5164e55eeb2b89f19c0e5c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-11 11:31:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/f88b4c3130c84c0b8eaf7ef2b18ef3e5 : POST_FAILURE in 1m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b06dd5432b9dabe39d9a01438a1ef6662a5fdbbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 11:34:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e936d701fdd94ba6b964cfb5062b5d9e : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"2e4c1a327b2fbe878d6ced76a40e65a93f3219db","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-11 11:42:09.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d4f0342b090a8b05fcc7da44058e7f982416248f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-11 11:44:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/30f889f215d14f90900d3bdd8e0403ea : POST_FAILURE in 1m 28s","accounts_in_message":[],"_revision_number":3},{"id":"49b21c7b9a0972877561cdc78d42dc1733506a8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 11:47:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3ee969364acc4a66ba9b4260ffa3f1c9 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"12c4b8d19a3cb2bb6f3746c125dc20e8789ae6f9","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-02-11 11:58:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1770cae934a7e01f6dd68a153bca6d2ea88a4dd5","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-11 12:00:35.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"37c3ebe341c888caf3363f046d70c238b89437c0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-11 12:03:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/68fbb14ad9fc463aaadf4f9ce3118574 : POST_FAILURE in 1m 15s","accounts_in_message":[],"_revision_number":4},{"id":"9dba4d9cc9586f1553bc5ae07e90e318c422854b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 12:09:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0b75b372977e464bb69dc41366ed4492 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":4},{"id":"43bc28f72d170e3adbb360fc08c53f6fe1540abb","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-11 12:09:21.000000000","message":"Patch Set 4:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":4},{"id":"4b4457c5bb76c7115cd46da746140f3a56b6f624","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-11 12:11:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/89feb1ec6cc9472e9e7e52e69907fbab : POST_FAILURE in 1m 16s","accounts_in_message":[],"_revision_number":4},{"id":"5c1beaf6182a065323ba0c27b0009da804ce6735","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-11 14:49:26.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"47afb23e111986f6240826be6cffd16520be2c8e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-11 14:51:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/19851a28780d40cc8a499f199ebbe85a : POST_FAILURE in 1m 12s","accounts_in_message":[],"_revision_number":5},{"id":"6c5372c52158cfa3de63d1a6efc1295215929e95","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-11 14:56:27.000000000","message":"Patch Set 5:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":5},{"id":"ae1906e6df2299685f80e370a182428c0cd823e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 14:59:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/14a035782286421697c50212f3b656a4 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":5},{"id":"186e8537a8ac5759e15e4e2ca1cebc17ed3ab180","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-11 15:02:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/1bc6eb07778c41b4bd553782cec5df1a : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":5},{"id":"eea38d5fa559142761c6d1cd2b226f539e4e7730","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-02-11 15:28:57.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5a6fe29f88f9d8755ac552f499866a3ddae7b75d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:29:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ac4b5cb24bd4aa27ab5be1586acf3aae77783e11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:35:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e1d24f709108429a851a110a010743b1 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":5},{"id":"ef9c37dbedb65b4556765e181f1200ab76b1953c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:35:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"22338e93eb5e9ff3f86860b8e372d767c55b9a74","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-03-13 15:53:45.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 955aaf9818a68b694ddfcdedd4c055b56b0838aa","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6c609c0fa5054ae0e12311825412e8515101b176","revisions":{"d8a8e6558c160993c2dcfa30ff4cb6deb870714a":{"kind":"REWORK","_number":1,"created":"2020-02-11 09:55:57.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/88/707088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/707088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/1"}}},"commit":{"parents":[{"commit":"da7f5a1ed073a0908d8010598d55bba6bbbc94e0","subject":"Merge \"mox3: convert to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/da7f5a1ed073a0908d8010598d55bba6bbbc94e0"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-02-11 09:55:43.000000000","tz":60},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-02-11 09:55:43.000000000","tz":60},"subject":"debtcollector: convert to python3","message":"debtcollector: convert to python3\n\nChange-Id: Ie28a92823b0a5340789e19860e5d01938807c9fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d8a8e6558c160993c2dcfa30ff4cb6deb870714a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d8a8e6558c160993c2dcfa30ff4cb6deb870714a"}]},"branch":"refs/heads/master"},"42253d04406232a84f60ed535b62c21c36aaf63d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-11 11:29:39.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/88/707088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/707088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/2"}}},"commit":{"parents":[{"commit":"da7f5a1ed073a0908d8010598d55bba6bbbc94e0","subject":"Merge \"mox3: convert to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/da7f5a1ed073a0908d8010598d55bba6bbbc94e0"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-02-11 09:55:43.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2020-02-11 11:29:37.000000000","tz":0},"subject":"debtcollector: convert to python3","message":"debtcollector: convert to python3\n\nChange-Id: Ie28a92823b0a5340789e19860e5d01938807c9fa\nDepends-On: https://review.rdoproject.org/r/24896\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/42253d04406232a84f60ed535b62c21c36aaf63d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/42253d04406232a84f60ed535b62c21c36aaf63d"}]},"branch":"refs/heads/master"},"6fde03229ecd5ecbbb59f64ff992b29fb97e4fab":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-11 11:42:09.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/88/707088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/707088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/3"}}},"commit":{"parents":[{"commit":"da7f5a1ed073a0908d8010598d55bba6bbbc94e0","subject":"Merge \"mox3: convert to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/da7f5a1ed073a0908d8010598d55bba6bbbc94e0"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-02-11 09:55:43.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2020-02-11 11:42:06.000000000","tz":0},"subject":"debtcollector: convert to python3","message":"debtcollector: convert to python3\n\nChange-Id: Ie28a92823b0a5340789e19860e5d01938807c9fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6fde03229ecd5ecbbb59f64ff992b29fb97e4fab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6fde03229ecd5ecbbb59f64ff992b29fb97e4fab"}]},"branch":"refs/heads/master"},"d330f8fbff03b38cbfdca35c6dcae974a9c57eae":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-02-11 12:00:35.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/88/707088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/707088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/4"}}},"commit":{"parents":[{"commit":"da7f5a1ed073a0908d8010598d55bba6bbbc94e0","subject":"Merge \"mox3: convert to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/da7f5a1ed073a0908d8010598d55bba6bbbc94e0"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-02-11 09:55:43.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2020-02-11 12:00:33.000000000","tz":0},"subject":"debtcollector: convert to python3","message":"debtcollector: convert to python3\n\nChange-Id: Ie28a92823b0a5340789e19860e5d01938807c9fa\nDepends-On: https://review.rdoproject.org/r/24897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d330f8fbff03b38cbfdca35c6dcae974a9c57eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d330f8fbff03b38cbfdca35c6dcae974a9c57eae"}]},"branch":"refs/heads/master"},"6c609c0fa5054ae0e12311825412e8515101b176":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-02-11 14:49:26.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/88/707088/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/707088/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/707088/5"}}},"commit":{"parents":[{"commit":"da7f5a1ed073a0908d8010598d55bba6bbbc94e0","subject":"Merge \"mox3: convert to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/da7f5a1ed073a0908d8010598d55bba6bbbc94e0"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-02-11 09:55:43.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2020-02-11 14:49:24.000000000","tz":0},"subject":"debtcollector: convert to python3","message":"debtcollector: convert to python3\n\nChange-Id: Ie28a92823b0a5340789e19860e5d01938807c9fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6c609c0fa5054ae0e12311825412e8515101b176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6c609c0fa5054ae0e12311825412e8515101b176"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
