)]}'
{"id":"openstack%2Frpm-packaging~707858","triplet_id":"openstack%2Frpm-packaging~master~I9f85e897bff5e9f77bf366f60e1ef5aa8c1c25ed","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"I9f85e897bff5e9f77bf366f60e1ef5aa8c1c25ed","subject":"designate: update dependencies (edgegrid-python is now needed)","status":"MERGED","created":"2020-02-14 15:20:40.000000000","updated":"2020-02-22 11:56:19.000000000","submitted":"2020-02-22 11:56:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707858-1582372579153-5f1f21a0","meta_rev_id":"2f36ab02b4252baa15b04980661676781b5f3f38","_number":707858,"virtual_id_number":707858,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2020-02-22 11:16:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-22 11:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-20 13:43:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2020-02-21 16:17:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-20 15:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":2,"date":"2020-02-21 13:53:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-22 11:51:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 13:43:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-20 15:01:42.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-02-21 13:53:10.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2020-02-21 16:17:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-22 11:16:46.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-22 11:56:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"165e120d89dace815dd6bce4e9ef47b88d7a5128","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-02-14 15:20:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a91b8737fd1990b7b20e5876d01f2035f56c6aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 15:33:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4aa906be3afb4125b87b3192c6217e8e : FAILURE in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a784f0155180d0c3a1314a73003dbbb7713c5694","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-14 17:36:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-fedora https://review.rdoproject.org/zuul/build/e97745de2e1c496d842bceb6ffeec181 : FAILURE in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b99dbfbec59b68580ae092174f1db008f016eb6f","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-14 21:39:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-6289d050b84d34c9816e6afcc1e7ba9a0a919e93 : FAILURE in 6h 18m 01s","accounts_in_message":[],"_revision_number":1},{"id":"cf6a0e68687f9755a85c439af72b0d5e261a5766","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-20 07:22:45.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM\n\nhttps://github.com/openstack/designate/commit/318b8d03193d3d9c26842528e9eaf25814df3ba9","accounts_in_message":[],"_revision_number":1},{"id":"3a913f142c8d04818acbc50157fa6ab27bc122f1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-20 07:25:36.000000000","message":"Patch Set 1:\n\nNot sure to seen why linter warn...\n\n```\n2020-02-14 15:31:54.566524 | ubuntu-bionic |  Requires:       python3-WebOb\n2020-02-14 15:31:54.566558 | ubuntu-bionic |  Requires:       python3-Werkzeug\n2020-02-14 15:31:54.566589 | ubuntu-bionic |  Requires:       python3-debtcollector\n2020-02-14 15:31:54.566612 | ubuntu-bionic | -Requires:       python3-edgegrid-python\n2020-02-14 15:31:54.566630 | ubuntu-bionic |  Requires:       python3-designateclient\n2020-02-14 15:31:54.566649 | ubuntu-bionic |  Requires:       python3-dnspython\n2020-02-14 15:31:54.566667 | ubuntu-bionic | +Requires:       python3-edgegrid-python\n2020-02-14 15:31:54.566685 | ubuntu-bionic |  Requires:       python3-eventlet\n2020-02-14 15:31:54.566703 | ubuntu-bionic |  Requires:       python3-futurist\n2020-02-14 15:31:54.566723 | ubuntu-bionic |  Requires:       python3-greenlet\n2020-02-14 15:31:54.575345 | ubuntu-bionic | ERROR: InvocationError for command /home/zuul/src/opendev.org/openstack/rpm-packaging/tools/run_speccleaner.sh . (exited with code 1)\n```","accounts_in_message":[],"_revision_number":1},{"id":"70fb7d75d19b809855f2be78c10896e7c588c143","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-02-20 13:37:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c62e1ed1795bb4746c05f59ab993fcbc777f072d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 13:43:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/e957d0872b4746c2a84aac86a4c40034 : FAILURE in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0d5c3f8b52c24aaa68fb68a41046a7793639a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 13:47:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ad0d0cdfdd864b18a0d112b869272b32 : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":2},{"id":"76c251cd89b7b6dc90a7df696859e94101ed4c2d","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-20 14:09:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-b7a1103e911f88e7b996a27dae9b17e97474511f : FAILURE in 31m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5e82f0935f8cf276c4a7107a79798c7c48ddc3bc","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-20 14:33:56.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"3e8ae935e83d8020863183ab8e61215dbe29b99a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 14:41:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/622e41ac1ebd47e393c3fea49ecf74ba : FAILURE in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"93d7ea36419fa46a564e3fc30fe612131cb6e8ff","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-20 14:44:27.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"3a226808227689e29fb884aadaf26b7b4497b4e2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 14:59:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/a7e08e267f8549a48b703078f3b557a7 : SUCCESS in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0a8a9a807e703fb38d302ed68a067edc4c02d11","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-20 15:01:36.000000000","message":"Patch Set 2:\n\nNice, so we have all dependencies in RDO now. We just had to add python3-edgegrid and python3-ndg_httpsclient","accounts_in_message":[],"_revision_number":2},{"id":"9dc0dd7eaae18be5d9a9ccd9d96bbee9d6d9177c","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-20 15:01:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0d7f988acbdf4b4afb53c18eef35d710098db288","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-02-21 13:53:10.000000000","message":"Patch Set 2: Code-Review+2\n\nSUSE CI needs https://build.opensuse.org/request/show/777989","accounts_in_message":[],"_revision_number":2},{"id":"25d5af1da894c6ff07dc3a106bd63448af4806d5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-21 16:17:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3c4b7b8db2870ad755db8ce41443e785bb903143","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-02-22 10:03:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5b1ff3c128e600be42bb0311d960f3681cc07e81","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-22 10:13:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/3b615cc71a994786a8a7e7c5c2d22785 : SUCCESS in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22f95d5cbf16418ba0e2b0aec584e930c871f956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 10:13:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ebe3f3c1518442c18303c4bcc921eb56 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b3ec77b0f047691a916e163b446e454e2f62e922","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-22 11:16:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-5eee20b1527e49805376db98d878920cb3e18212 : SUCCESS in 1h 12m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b6422da2d56984bf15c1e713faf16b465e82743b","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-02-22 11:51:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"edee79c5de57df1d2b2aeed82cee49e3abe9f97c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 11:51:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b9ad6b014e7974fabfce17821cde74d56c71833","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 11:56:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b731304a6fa746dda506901147880144 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"2f36ab02b4252baa15b04980661676781b5f3f38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 11:56:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"04d1f9a778fe616581f6521ca79d9af29fd2c619","revisions":{"6289d050b84d34c9816e6afcc1e7ba9a0a919e93":{"kind":"REWORK","_number":1,"created":"2020-02-14 15:20:40.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/58/707858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/58/707858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/58/707858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/58/707858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/58/707858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/58/707858/1"}}},"commit":{"parents":[{"commit":"d5fd238dc4419036151fc1a16f4267deb931ff36","subject":"Merge \"openstacksdk: Temporarily skip unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d5fd238dc4419036151fc1a16f4267deb931ff36"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-02-14 15:20:21.000000000","tz":60},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-02-14 15:20:21.000000000","tz":60},"subject":"designate: update dependencies (edgegrid-python is now needed)","message":"designate: update dependencies (edgegrid-python is now needed)\n\nChange-Id: I9f85e897bff5e9f77bf366f60e1ef5aa8c1c25ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6289d050b84d34c9816e6afcc1e7ba9a0a919e93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6289d050b84d34c9816e6afcc1e7ba9a0a919e93"}]},"branch":"refs/heads/master"},"04d1f9a778fe616581f6521ca79d9af29fd2c619":{"kind":"REWORK","_number":2,"created":"2020-02-20 13:37:35.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/58/707858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/58/707858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/58/707858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/58/707858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/58/707858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/58/707858/2"}}},"commit":{"parents":[{"commit":"d5fd238dc4419036151fc1a16f4267deb931ff36","subject":"Merge \"openstacksdk: Temporarily skip unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d5fd238dc4419036151fc1a16f4267deb931ff36"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-02-14 15:20:21.000000000","tz":60},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-02-20 13:37:22.000000000","tz":60},"subject":"designate: update dependencies (edgegrid-python is now needed)","message":"designate: update dependencies (edgegrid-python is now needed)\n\nChange-Id: I9f85e897bff5e9f77bf366f60e1ef5aa8c1c25ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/04d1f9a778fe616581f6521ca79d9af29fd2c619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/04d1f9a778fe616581f6521ca79d9af29fd2c619"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
