)]}'
{"id":"openstack%2Frpm-packaging~708888","triplet_id":"openstack%2Frpm-packaging~master~Ifb5c249788c2384bf70d25e7c3c947fbc347a984","project":"openstack/rpm-packaging","branch":"master","topic":"remove-python2","hashtags":[],"change_id":"Ifb5c249788c2384bf70d25e7c3c947fbc347a984","subject":"heat-agents: fix shebangs","status":"MERGED","created":"2020-02-20 14:04:33.000000000","updated":"2020-03-16 17:38:44.000000000","submitted":"2020-02-21 14:00:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708888-1582293629730-25e6796c","meta_rev_id":"fd7cebe2a1dd6e9dd7f614bbc08c7a8e5a362b7e","_number":708888,"virtual_id_number":708888,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2020-02-20 14:35:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":6679,"name":"Ryota Hashimoto","email":"hashimoto.ryt@gmail.com","username":"hashimotor"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-21 14:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-20 14:26:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-20 14:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":6679,"name":"Ryota Hashimoto","email":"hashimoto.ryt@gmail.com","username":"hashimotor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-21 13:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"date":"2020-02-21 13:47:13.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2020-02-21 13:47:14.000000000","_account_id":6679,"name":"Ryota Hashimoto","email":"hashimoto.ryt@gmail.com","username":"hashimotor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-21 13:51:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":6679,"name":"Ryota Hashimoto","email":"hashimoto.ryt@gmail.com","username":"hashimotor"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 14:26:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-20 14:29:39.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-02-20 14:35:09.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-21 13:47:13.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-02-21 13:47:14.000000000","updated_by":{"_account_id":6679,"name":"Ryota Hashimoto","email":"hashimoto.ryt@gmail.com","username":"hashimotor"},"reviewer":{"_account_id":6679,"name":"Ryota Hashimoto","email":"hashimoto.ryt@gmail.com","username":"hashimotor"},"state":"REVIEWER"},{"updated":"2020-02-21 13:51:16.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2020-02-21 14:00:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04ed62854bfc155ffc0a33aa7ea490124e29da05","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-20 14:04:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24a7ebffe50287c7c095f88316e143859922d83f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 14:13:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/2c0dd18915dc4af7885b56915f4059a9 : SUCCESS in 7m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04c4a85884db5c01e6b6284c955e785b4fa582f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 14:14:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c465d28ea4934efd9746278a26c7d02c : FAILURE in 8m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1e6cc3df53cacd2f680814c5a8022fa16590a10d","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-20 14:17:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"febfcf73697ad21d2e4f49cd2f7d78493cfbfc21","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-20 14:18:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"de7664c1035534dbb7c0f13e32730adb3e022835","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 14:26:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/bbf5fc01cbad4099afe1865c126c27a4 : SUCCESS in 7m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1c76d9dc369952535fc75c0071d1f10e46b563b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 14:27:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dcc733e025824cdfa6c5ee9532111621 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"0bf913009a7a54e44296a6df6b277d2bcaeff35e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-02-20 14:29:39.000000000","message":"Patch Set 3: Code-Review+2\n\nI wonder if we can fix that within heat-agents itself. this seems lik ea really bad default to be based on env.","accounts_in_message":[],"_revision_number":3},{"id":"7bdab3914cbb3336d634fd9362dc7fc662a63387","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-20 14:35:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-8672c8b8795cd2b26216447f982ac5a9662d8812 : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":3},{"id":"98ead61d41f73b507355c8c982b82c1026c178a5","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-02-21 13:51:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"031d68fc36e6c70bbbc7fe7d85c813fcab0f204e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 13:51:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a167868af699b2d0f1668a84471d6d8b028d4379","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 14:00:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a9d1bfee0e104bcf95257d59602e7495 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":3},{"id":"1dde98e4fce88a53c7fa3723369cc8585e8279be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 14:00:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"aa7041fdc7e6494058850fbdc3f0f14d7b3f974d","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-03-16 17:38:44.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 86cac5a1da9a3d2df5fb4c84beffc4a201830eee","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8672c8b8795cd2b26216447f982ac5a9662d8812","revisions":{"941734878ab8e7e8859560876123e81b95e88711":{"kind":"REWORK","_number":1,"created":"2020-02-20 14:04:33.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/88/708888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/708888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/1"}}},"commit":{"parents":[{"commit":"b471985b5c4f2b9f8fd7f27fc7a46706c96850bb","subject":"Merge \"heat-dashboard: fix python3 reqs for -test subpackage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b471985b5c4f2b9f8fd7f27fc7a46706c96850bb"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-02-20 14:03:45.000000000","tz":60},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-02-20 14:03:45.000000000","tz":60},"subject":"heat-agents: fix shebangs","message":"heat-agents: fix shebangs\n\nWe need to make sure the python shebangs using \"/usr/bin/env python\"\nare changed to \"/usr/bin/python3\", otherwise the build process will\nfail on CentOS 8.\n\nChange-Id: Ifb5c249788c2384bf70d25e7c3c947fbc347a984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/941734878ab8e7e8859560876123e81b95e88711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/941734878ab8e7e8859560876123e81b95e88711"}]},"branch":"refs/heads/master"},"7affdd47ab8f897f22f9a2bc071846d43614e3ac":{"kind":"REWORK","_number":2,"created":"2020-02-20 14:17:52.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/88/708888/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/708888/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/2"}}},"commit":{"parents":[{"commit":"b471985b5c4f2b9f8fd7f27fc7a46706c96850bb","subject":"Merge \"heat-dashboard: fix python3 reqs for -test subpackage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b471985b5c4f2b9f8fd7f27fc7a46706c96850bb"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-02-20 14:03:45.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2020-02-20 14:17:50.000000000","tz":0},"subject":"heat-agents: fix shebangs","message":"heat-agents: fix shebangs\n\nWe need to make sure the python shebangs using \"/usr/bin/env python\"\nare changed to \"/usr/bin/python3\", otherwise the build process will\nfail on CentOS 8.\n\nChange-Id: Ifb5c249788c2384bf70d25e7c3c947fbc347a984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7affdd47ab8f897f22f9a2bc071846d43614e3ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7affdd47ab8f897f22f9a2bc071846d43614e3ac"}]},"branch":"refs/heads/master"},"8672c8b8795cd2b26216447f982ac5a9662d8812":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-20 14:18:51.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/88/708888/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/88/708888/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/88/708888/3"}}},"commit":{"parents":[{"commit":"b5d4545f4ba3c169409e2a65d0a998f44887f3fe","subject":"Merge \"update oslo.reports to 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/b5d4545f4ba3c169409e2a65d0a998f44887f3fe"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-02-20 14:03:45.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2020-02-20 14:18:51.000000000","tz":0},"subject":"heat-agents: fix shebangs","message":"heat-agents: fix shebangs\n\nWe need to make sure the python shebangs using \"/usr/bin/env python\"\nare changed to \"/usr/bin/python3\", otherwise the build process will\nfail on CentOS 8.\n\nChange-Id: Ifb5c249788c2384bf70d25e7c3c947fbc347a984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8672c8b8795cd2b26216447f982ac5a9662d8812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8672c8b8795cd2b26216447f982ac5a9662d8812"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
