)]}'
{"id":"openstack%2Frpm-packaging~709037","triplet_id":"openstack%2Frpm-packaging~master~Ifbeb2df6d579dd166d90da2b90aedfed50f7260e","project":"openstack/rpm-packaging","branch":"master","topic":"keystoneauth1","hashtags":[],"change_id":"Ifbeb2df6d579dd166d90da2b90aedfed50f7260e","subject":"update keystoneauth1 for stable","status":"ABANDONED","created":"2020-02-21 03:33:56.000000000","updated":"2020-02-21 09:25:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"060430280bebe320274bc173074eb180dff39c1f","_number":709037,"virtual_id_number":709037,"owner":{"_account_id":27462,"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","username":"PengZhao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2020-02-21 03:39:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"date":"2020-02-21 09:08:16.000000000","_account_id":27462,"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","username":"PengZhao"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-21 03:39:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-21 03:40:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27462,"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","username":"PengZhao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-02-21 08:49:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27462,"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","username":"PengZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27462,"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","username":"PengZhao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-21 03:39:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-21 03:39:31.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-21 03:40:16.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-21 08:49:05.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"}],"messages":[{"id":"24b18fcec91a9241d95171522e5bf976c411b904","author":{"_account_id":27462,"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","username":"PengZhao"},"date":"2020-02-21 03:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cf638f88e03de3c064db75b1a2c72b71d1cb006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 03:39:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1da29893406b442ba00f2bac5be8eae9 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2660bde6b7b0790ea5032f54a801f9c33be35166","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-21 03:39:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-e8c49489ef4298b363f84bb23a65ac72b4427670 : FAILURE in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"052f56312178635c0b8fbffc465cc8dda2792926","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-21 03:40:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/ff8f7366406e46e4ac785567f75aedd3 : FAILURE in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63f4fb2ec33091466270ba59d7d2a9c5493ce854","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-21 08:49:05.000000000","message":"Patch Set 1: Code-Review-1\n\nThis change is proposed for the master branch, and means using an earlier version. I guess you meant to do it on stable/train?","accounts_in_message":[],"_revision_number":1},{"id":"07d4f2e305c0b786710ec44203f40f1e5e859781","author":{"_account_id":27462,"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","username":"PengZhao"},"date":"2020-02-21 09:08:16.000000000","message":"Patch Set 1:\n\nsorry!I have a doubt. I should commit this change to the stable/train branch, right?","accounts_in_message":[],"_revision_number":1},{"id":"cb0e3e5265c9fc2dbd590135b915dc7d48387f1d","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-02-21 09:16:08.000000000","message":"Patch Set 1:\n\n\u003e sorry!I have a doubt. I should commit this change to the\n \u003e stable/train branch, right?\n\nRight, that\u0027s the branch where this change should be proposed. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"44940fff9d51870c985a5ccb947eaec45958c245","author":{"_account_id":27462,"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","username":"PengZhao"},"date":"2020-02-21 09:25:52.000000000","message":"Abandoned\n\ncommit to a wrong branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8c49489ef4298b363f84bb23a65ac72b4427670","revisions":{"e8c49489ef4298b363f84bb23a65ac72b4427670":{"kind":"REWORK","_number":1,"created":"2020-02-21 03:33:56.000000000","uploader":{"_account_id":27462,"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","username":"PengZhao"},"ref":"refs/changes/37/709037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/37/709037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/37/709037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/37/709037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/37/709037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/37/709037/1"}}},"commit":{"parents":[{"commit":"c9e9148d03b8a374c387ff2489f16dd55af24ea2","subject":"Merge \"update constraint for python-swiftclient to new release 3.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/c9e9148d03b8a374c387ff2489f16dd55af24ea2"}]}],"author":{"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","date":"2020-02-21 03:33:40.000000000","tz":480},"committer":{"name":"Peng Zhao","email":"zhaop33@chinaunicom.cn","date":"2020-02-21 03:33:40.000000000","tz":480},"subject":"update keystoneauth1 for stable","message":"update keystoneauth1 for stable\n\nChange-Id: Ifbeb2df6d579dd166d90da2b90aedfed50f7260e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/e8c49489ef4298b363f84bb23a65ac72b4427670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/e8c49489ef4298b363f84bb23a65ac72b4427670"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
