)]}'
{"id":"openstack%2Frpm-packaging~709429","triplet_id":"openstack%2Frpm-packaging~master~I6592a4232eee8cdbda6eb709ded6c28bb4571b51","project":"openstack/rpm-packaging","branch":"master","topic":"oslo.service","hashtags":[],"change_id":"I6592a4232eee8cdbda6eb709ded6c28bb4571b51","subject":"update constraint for oslo.service to new release 2.0.0","status":"ABANDONED","created":"2020-02-24 08:42:21.000000000","updated":"2020-02-25 01:35:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c7590646a0426858d3b992e1847c640df46cfc4b","_number":709429,"virtual_id_number":709429,"owner":{"_account_id":26571,"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","username":"sobjustin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2020-02-24 08:55:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-24 08:47:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-24 08:50:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-02-24 10:35:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 08:47:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-24 08:50:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-24 08:55:19.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-24 10:35:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"db2c24d53f72bbe8a9212f39470d0e77412e4038","author":{"_account_id":26571,"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","username":"sobjustin"},"date":"2020-02-24 08:42:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcd24f3f2efb0718a39220ed6761f7a1f6cdf774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 08:47:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/55332c183ff649e198471886138c232d : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"922491cc01292eb7e55785f05c69fc60360b19fe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 08:50:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/00b57020aca84339bd3b4586da40ca57 : SUCCESS in 7m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45b7b9567c80e63d3f980396c377bee5af77f6d0","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-24 08:55:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-a9fc47ecb913d3d9db131c27dc280f442768fe9c : FAILURE in 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a7edec6d210dbf494379c696edc80f0c8220447b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-24 10:35:23.000000000","message":"Patch Set 1: Code-Review-1\n\nAlready addressed in https://review.opendev.org/#/c/709315/","accounts_in_message":[],"_revision_number":1},{"id":"0b8100af72cbdcb95753ac14d51d61ec902e8f0b","author":{"_account_id":26571,"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","username":"sobjustin"},"date":"2020-02-25 01:35:35.000000000","message":"Abandoned\n\nDuplicated","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9fc47ecb913d3d9db131c27dc280f442768fe9c","revisions":{"a9fc47ecb913d3d9db131c27dc280f442768fe9c":{"kind":"REWORK","_number":1,"created":"2020-02-24 08:42:21.000000000","uploader":{"_account_id":26571,"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","username":"sobjustin"},"ref":"refs/changes/29/709429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/29/709429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/29/709429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/29/709429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/29/709429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/29/709429/1"}}},"commit":{"parents":[{"commit":"33d4dc817cfa27f5c92aa8ee7724bc6f6b1e1038","subject":"Merge \"update python-barbicanclient to new release 4.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/33d4dc817cfa27f5c92aa8ee7724bc6f6b1e1038"}]}],"author":{"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","date":"2020-02-24 08:41:34.000000000","tz":480},"committer":{"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","date":"2020-02-24 08:41:34.000000000","tz":480},"subject":"update constraint for oslo.service to new release 2.0.0","message":"update constraint for oslo.service to new release 2.0.0\n\nDepends-on:I79d0105537898fb82a4cfae38b101797164d53d6\nChange-Id: I6592a4232eee8cdbda6eb709ded6c28bb4571b51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a9fc47ecb913d3d9db131c27dc280f442768fe9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a9fc47ecb913d3d9db131c27dc280f442768fe9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
