)]}'
{"id":"openstack%2Frpm-packaging~709931","triplet_id":"openstack%2Frpm-packaging~stable%2Ftrain~Ib1b155646db34b0be5c9bc3d785bc4eef8449360","project":"openstack/rpm-packaging","branch":"stable/train","topic":"oslo.utils","hashtags":[],"change_id":"Ib1b155646db34b0be5c9bc3d785bc4eef8449360","subject":"update constraint for oslo.utils to new release 3.41.4","status":"MERGED","created":"2020-02-26 02:08:20.000000000","updated":"2020-03-16 09:29:48.000000000","submitted":"2020-03-16 09:29:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709931-1584350988023-3b0434b8","meta_rev_id":"85ba4fb7aa68abca140c780070aa6d43d6641647","_number":709931,"virtual_id_number":709931,"owner":{"_account_id":26571,"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","username":"sobjustin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-16 09:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-16 08:59:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-16 08:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2020-03-16 08:52:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-16 08:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-16 09:20:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 14:43:54.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-03-10 09:04:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-10 19:05:25.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-16 08:59:43.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-16 09:20:01.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-03-16 09:29:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7f01a00543d86e294c80bf483b853d1b14c7af4","author":{"_account_id":26571,"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","username":"sobjustin"},"date":"2020-02-26 02:08:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa27fce61e1a574a87f19b70a1041febcd95cb67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 02:15:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd0931a51a574223b25ed7f0c2f4a230 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5a6e3180fdc0f8c420f22a7b91d7122e06a31655","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-26 02:17:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8-nv https://review.rdoproject.org/zuul/build/6a852711083f453bb3c07eab7ba1a347 : FAILURE in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"404ecd00b4fb1f3a420bf34f05970fdcf416d23c","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-26 07:41:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-9d5a17fb5451107c0cad75310d12d4c4abfc8536 : FAILURE in 5h 33m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6951e5c8d2c0b00f8f3f86853a9089cb0806eb0a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-01 20:00:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3afba7780294084f1ada0c5670c4ecf6e002f6ee","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-01 20:01:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ad6d14c5bfff91ab0971e2a243d42f57626d8ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 20:06:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2c2cefd072e249fa934d14f80b7a7ec6 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3a38d57dbcbf3347b75cb6e946f459bd9e90ce6b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-01 20:09:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8-nv https://review.rdoproject.org/zuul/build/a02f78e6a1374101b762d18dbd9100b4 : FAILURE in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5837edc91c0d205127733de49f8d8de7d9ad4922","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-03-02 10:33:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-a0b4858eb956791b73d37499b2c66be30d42d73e : FAILURE in 12m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ac2e5b31c282f5cb3d3698a1cc6ec97ca49a7cf1","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-03-05 14:43:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a9db762ced6c54cf02006845ece832adfb292a77","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-10 09:04:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"002d77f0222c165cdc98c4040a4f3c59e0cbdf77","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-10 17:22:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"df52cd0efb54d6391d28e11f1d71cb004d1822a4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-10 17:30:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8-nv https://review.rdoproject.org/zuul/build/b53bc90c31444a1c86c9317fccbcce2f : FAILURE in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1411d416eb04edb6515765cff2eec3b44b4c1591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 17:33:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2d2bcf7633e9448582b530f10ac0b961 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":3},{"id":"7a055776cf787d927d8f82349300f548d104d001","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-03-10 19:05:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-c8e744494906732fe91b351b9472550dca7eb5c7 : FAILURE in 1h 29m 24s","accounts_in_message":[],"_revision_number":3},{"id":"61af2a9f494dda100049b80ce6385664c2dacd7e","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-03-16 08:52:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f76387cd378d3b30da6412782fa89e0f1cb349c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 08:58:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/29a497abcf0348a1bf250f0df1e103fe : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":4},{"id":"dc2e2d6eaf8d29e69b2a7876b891569850be7ab7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-16 08:59:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8-nv https://review.rdoproject.org/zuul/build/a75a52dd53984ae790fa54ef7489cd0c : SUCCESS in 6m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"687b5f8ba279e6603b15b6d73651f1acbf921a5c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-16 09:20:01.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"002d3b3693e4fc76a20b542552f86547b3f13b49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 09:20:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b6d282b710bc9b92edbbbf527c47c3f8f24bca34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 09:29:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5dadebe25bf64085971c8ff34114166d : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":4},{"id":"3b94cd2234943b1ecb91180d93b0326130e902bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 09:29:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0fb2ed5cf45108b6a23fb1e8b302b94a76a60239","revisions":{"9d5a17fb5451107c0cad75310d12d4c4abfc8536":{"kind":"REWORK","_number":1,"created":"2020-02-26 02:08:20.000000000","uploader":{"_account_id":26571,"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","username":"sobjustin"},"ref":"refs/changes/31/709931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/31/709931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/1"}}},"commit":{"parents":[{"commit":"77982dfcca0ca8a3a87b0f15e24e9af5e989083e","subject":"Merge \"add X.509 certificate check plugin for monasca-agent\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/77982dfcca0ca8a3a87b0f15e24e9af5e989083e"}]}],"author":{"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","date":"2020-02-26 02:06:39.000000000","tz":480},"committer":{"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","date":"2020-02-26 02:08:10.000000000","tz":480},"subject":"update constraint for oslo.utils to new release 3.41.4","message":"update constraint for oslo.utils to new release 3.41.4\n\nDepends-on:I39e3371ce613d2a5a1480b789d099d7772d66463\nChange-Id: Ib1b155646db34b0be5c9bc3d785bc4eef8449360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9d5a17fb5451107c0cad75310d12d4c4abfc8536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9d5a17fb5451107c0cad75310d12d4c4abfc8536"}]},"branch":"refs/heads/stable/train"},"a0b4858eb956791b73d37499b2c66be30d42d73e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-01 20:00:59.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/31/709931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/31/709931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/2"}}},"commit":{"parents":[{"commit":"f758639f286fcacf2034342b6e32110fcfe041b1","subject":"Merge \"python-senlinclient: convert to python3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/f758639f286fcacf2034342b6e32110fcfe041b1"}]}],"author":{"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","date":"2020-02-26 02:06:39.000000000","tz":480},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-03-01 20:00:59.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 3.41.4","message":"update constraint for oslo.utils to new release 3.41.4\n\nDepends-on:I39e3371ce613d2a5a1480b789d099d7772d66463\nChange-Id: Ib1b155646db34b0be5c9bc3d785bc4eef8449360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a0b4858eb956791b73d37499b2c66be30d42d73e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a0b4858eb956791b73d37499b2c66be30d42d73e"}]},"branch":"refs/heads/stable/train"},"c8e744494906732fe91b351b9472550dca7eb5c7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-10 17:22:37.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/31/709931/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/31/709931/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/3"}}},"commit":{"parents":[{"commit":"0eb45fbc611775b62520e13f955825c370776bf2","subject":"Merge \"python-cloudkittyclient: convert to python3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0eb45fbc611775b62520e13f955825c370776bf2"}]}],"author":{"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","date":"2020-02-26 02:06:39.000000000","tz":480},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-03-10 17:22:37.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 3.41.4","message":"update constraint for oslo.utils to new release 3.41.4\n\nDepends-on:I39e3371ce613d2a5a1480b789d099d7772d66463\nChange-Id: Ib1b155646db34b0be5c9bc3d785bc4eef8449360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/c8e744494906732fe91b351b9472550dca7eb5c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/c8e744494906732fe91b351b9472550dca7eb5c7"}]},"branch":"refs/heads/stable/train"},"0fb2ed5cf45108b6a23fb1e8b302b94a76a60239":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-16 08:52:36.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/31/709931/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/31/709931/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/31/709931/4"}}},"commit":{"parents":[{"commit":"8a91528621759f39c05544cc9b164393c201e377","subject":"Merge \"pymod2pkg: convert to python3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8a91528621759f39c05544cc9b164393c201e377"}]}],"author":{"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","date":"2020-02-26 02:06:39.000000000","tz":480},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2020-03-16 08:52:36.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 3.41.4","message":"update constraint for oslo.utils to new release 3.41.4\n\nDepends-on:I39e3371ce613d2a5a1480b789d099d7772d66463\nChange-Id: Ib1b155646db34b0be5c9bc3d785bc4eef8449360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0fb2ed5cf45108b6a23fb1e8b302b94a76a60239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0fb2ed5cf45108b6a23fb1e8b302b94a76a60239"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
