)]}'
{"id":"openstack%2Frpm-packaging~709932","triplet_id":"openstack%2Frpm-packaging~stable%2Ftrain~I0cae22dd09d827cf3447cdfe08d434863642ff52","project":"openstack/rpm-packaging","branch":"stable/train","topic":"oslo.config","hashtags":[],"change_id":"I0cae22dd09d827cf3447cdfe08d434863642ff52","subject":"update constraint for oslo.config to new release 6.11.2","status":"MERGED","created":"2020-02-26 02:15:45.000000000","updated":"2020-03-04 23:37:02.000000000","submitted":"2020-03-04 23:37:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709932-1583365022843-cb471cbf","meta_rev_id":"ef34c2cd9d45d61afa545694511ae435314056c6","_number":709932,"virtual_id_number":709932,"owner":{"_account_id":26571,"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","username":"sobjustin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2020-03-02 10:35:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 23:37:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-01 20:08:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-04 11:51:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-01 19:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-04 23:29:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-29 05:12:49.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2020-03-01 20:08:41.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-02 10:35:37.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-04 23:29:56.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-03-04 23:37:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6727b4903e68f777aded52324c59293033a5099f","author":{"_account_id":26571,"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","username":"sobjustin"},"date":"2020-02-26 02:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73b78057ec590f0126caa7678b0ec6bf3ff24590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 02:21:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b7bba45dd9704895b1d7c99d1de8fa95 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a7f57dd2a8090ee7d4fec2dbf16c6cc2c3b9eee9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-26 02:22:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8-nv https://review.rdoproject.org/zuul/build/1e8ed408d1bc40b7aaab976d5d562ff4 : FAILURE in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3d7033d3fac66a69ea8404c7b5c3d1f14d0556d","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-02-26 08:22:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-30e83a307da5d5d187e048cd2870e4c95dcc039b : FAILURE in 6h 06m 37s","accounts_in_message":[],"_revision_number":1},{"id":"169f294f432e7cf5fc49dc8f76a34ec1f8b4ccea","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-02-29 05:12:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"42b64a2937b6d20f1a98c061d69764bef9c9efea","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-01 19:59:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f3d93c433ea95df520ad690e67dea7d848c7b170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 20:05:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b4e7b7681879417abc3253fe8e18f8ae : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"80637f8c9fecede4d506384e5df6910c64097b2a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-01 20:08:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8-nv https://review.rdoproject.org/zuul/build/4fab982725ea421290e4a8ce8968d673 : FAILURE in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"40855bbd36b3c6ab550889563e121306ffb7db59","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-03-02 10:35:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-1b6079aa0ed5565f5b785e78e3446ea9f58f31dc : FAILURE in 14m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9c04c9ce098f1304af3b6fc44defa6b8c854df59","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-04 11:51:55.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d7e6bea7cc12f7071b6116a7c202a50ac80c69e9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-04 11:58:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8-nv https://review.rdoproject.org/zuul/build/2c44bfbe1a9741a8a6fdf8af58d46a63 : FAILURE in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e65e753b2906c931da60f5f0c668edff81688700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:02:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/becd09b3655347fdb6711470dce1a633 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c10c3725ff6713b75e331dc98bda6b6b46985946","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 13:39:05.000000000","message":"Patch Set 2:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-23be99fab1d02112707ee6a6f7954e220dc527d5 : FAILURE in 1h 46m 49s","accounts_in_message":[],"_revision_number":2},{"id":"32b5f3d3d106fbd138f0500c44b883a9ad1cb0e6","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-04 23:29:56.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"50625ed2885c96bf3f844d294261788660c0f1ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 23:30:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14a6a3a4cf0365ff6a925534ea6ecdc779681a0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 23:37:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e230abc6a9434868bdf6171282d084bd : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"009f9af6a7cd3c80b2acfb096b6bbf239ea81ae0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 23:37:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1b6079aa0ed5565f5b785e78e3446ea9f58f31dc","revisions":{"30e83a307da5d5d187e048cd2870e4c95dcc039b":{"kind":"REWORK","_number":1,"created":"2020-02-26 02:15:45.000000000","uploader":{"_account_id":26571,"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","username":"sobjustin"},"ref":"refs/changes/32/709932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/32/709932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/32/709932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/32/709932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/32/709932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/32/709932/1"}}},"commit":{"parents":[{"commit":"77982dfcca0ca8a3a87b0f15e24e9af5e989083e","subject":"Merge \"add X.509 certificate check plugin for monasca-agent\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/77982dfcca0ca8a3a87b0f15e24e9af5e989083e"}]}],"author":{"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","date":"2020-02-26 02:14:11.000000000","tz":480},"committer":{"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","date":"2020-02-26 02:14:11.000000000","tz":480},"subject":"update constraint for oslo.config to new release 6.11.2","message":"update constraint for oslo.config to new release 6.11.2\n\nDepends-on:I83393e38e9440b9cf0a17e2baef51f37e5a19700\nChange-Id: I0cae22dd09d827cf3447cdfe08d434863642ff52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/30e83a307da5d5d187e048cd2870e4c95dcc039b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/30e83a307da5d5d187e048cd2870e4c95dcc039b"}]},"branch":"refs/heads/stable/train"},"1b6079aa0ed5565f5b785e78e3446ea9f58f31dc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-01 19:59:56.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/32/709932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/32/709932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/32/709932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/32/709932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/32/709932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/32/709932/2"}}},"commit":{"parents":[{"commit":"f758639f286fcacf2034342b6e32110fcfe041b1","subject":"Merge \"python-senlinclient: convert to python3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/f758639f286fcacf2034342b6e32110fcfe041b1"}]}],"author":{"name":"Chang Liu","email":"liuchang77@chinaunicom.cn","date":"2020-02-26 02:14:11.000000000","tz":480},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-03-01 19:59:56.000000000","tz":0},"subject":"update constraint for oslo.config to new release 6.11.2","message":"update constraint for oslo.config to new release 6.11.2\n\nDepends-on:I83393e38e9440b9cf0a17e2baef51f37e5a19700\nChange-Id: I0cae22dd09d827cf3447cdfe08d434863642ff52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1b6079aa0ed5565f5b785e78e3446ea9f58f31dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1b6079aa0ed5565f5b785e78e3446ea9f58f31dc"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
