)]}'
{"id":"openstack%2Frpm-packaging~726406","triplet_id":"openstack%2Frpm-packaging~master~Ie80c8fbdc851cbace8c82f8c47f490898f5c4d6e","project":"openstack/rpm-packaging","branch":"master","topic":"lockdown_ip_command_in_sudoers","hashtags":[],"change_id":"Ie80c8fbdc851cbace8c82f8c47f490898f5c4d6e","subject":"Lockdown /bin/ip permissions for the monasca-agent","status":"MERGED","created":"2020-05-08 14:44:36.000000000","updated":"2020-05-15 10:34:13.000000000","submitted":"2020-05-15 10:34:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726406-1589538853783-15620326","meta_rev_id":"9f5f42567e7f13612dd9d99dd1b3533d02ad222d","_number":726406,"virtual_id_number":726406,"owner":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2020-05-14 19:50:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 10:34:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":-1,"date":"2020-05-14 18:54:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-14 18:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2020-05-14 18:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-15 10:27:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 07:28:37.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2020-05-11 18:39:02.000000000","updated_by":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"reviewer":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"state":"REVIEWER"},{"updated":"2020-05-14 18:54:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-14 19:50:05.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-15 10:27:50.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-05-15 10:34:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99a12870cc70616568bc52618b05e28f586497b6","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-05-08 14:44:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08e077bd41c3f2a24ffbe7c453face4b14a4eec4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-08 14:47:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/e9fb0f9b8a304d37bb3237523ed20715 : POST_FAILURE in 1m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0263fac0129dcb5af5589d157407d36a9adacc29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 14:51:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d06fc68cd091474ba07fffd1be8730be : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d9dc1e9e9e7590920dac3b69688b22b0b65cdd0b","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-05-08 15:39:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-00077266ae8d70e5d79eb65fdb8081150cbc7842 : FAILURE in 54m 16s","accounts_in_message":[],"_revision_number":1},{"id":"55d176990d4086ea9d6ef90460ab660d78a15995","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-05-08 22:14:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"54a38d69794ae25956331d66b03f9166828e4ede","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-05-11 07:28:37.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a21f4b0afddffa920af0f6cb3e7e81f4752278d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-11 07:31:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/21732dd5d1254cc3a1db0b16fa943f02 : POST_FAILURE in 2m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a83a0da1b5d604a35653fffa0a06917298a44db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 07:33:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3445a7e502c6470091168c5450c91bb6 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4171cbf83dbb06d4858843ece7d1beda84c299ec","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-05-11 08:00:15.000000000","message":"Patch Set 1:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-b06f62ea6a9b179c250dbb15cc08badf32321997 : FAILURE in 31m 19s","accounts_in_message":[],"_revision_number":1},{"id":"32ddc4837c34db451ba8766f21bc503aabd5f147","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-05-11 11:35:40.000000000","message":"Patch Set 1: Code-Review-1\n\nThis doesn\u0027t lock down anything as it still allows it to execute anything.","accounts_in_message":[],"_revision_number":1},{"id":"91451df8862e3325b437ae9a3ba864d144d115e3","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-05-11 15:40:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5903a1fb782c617560ce2210798cb70b793e33d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-11 15:42:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/fdd081ef52c64d58b642b5e65d4d6407 : POST_FAILURE in 1m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c4739b9f52ae55d14187301cd7e2d20734bd0684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 15:49:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7bcfd9d3f60a4c4f9ec2ef1317255acf : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"055824fac55e1f21d5b82f88eb1c11f1687aa335","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-05-11 15:57:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"660362a07c43d529a96444445ceb70418c213b2c","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-05-11 17:09:12.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nSorry for asking for this, seems it is not possible.","accounts_in_message":[],"_revision_number":2},{"id":"adb35c22964cdcac6f0b9c8ce3022fc78b233e8b","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-05-11 17:41:16.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e Sorry for asking for this, seems it is not possible.\n\nwe could do something like restrict it starts with qrouter \n\nhttps://github.com/openstack/monasca-agent/blob/master/monasca_agent/collector/checks_d/libvirt.py#L303","accounts_in_message":[],"_revision_number":2},{"id":"319261d3de2e097cabd469d7a1878658841c4ead","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-05-11 17:54:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4031a9d79597a8e86adaa7a78fe5950bebd3cfd9","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-05-11 17:56:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0ce74924e552ff623a8a91d8427607b6a9a04bed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-11 17:57:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/1a445d629ada431fbb6631116a0a5683 : POST_FAILURE in 1m 52s","accounts_in_message":[],"_revision_number":3},{"id":"67935c0cbc626ecfcba2b8cf1466144c532c59e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:02:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c99ae8b65e404980acc9b4dc714bb68f : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":3},{"id":"564401ad090dc71f31da0c89995b87298a2318f8","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-05-11 18:06:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"75fa47d1cc6550b0592ee5aa486ae3a23e355a02","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-11 18:09:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/7d36d6c4480140ec8634bd2b0a8b2d4f : POST_FAILURE in 2m 09s","accounts_in_message":[],"_revision_number":4},{"id":"ab727fa3859e26fca5e61c8b0fedfce80b649e30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:12:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8b1484c6b02348e7a28e852c54a595ac : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":4},{"id":"1618d33ce4ac964c867ac1a9ffae3a08c4eea8b0","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-05-11 18:35:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-b2a50ad21bae240e17a89b997b89c3a16528f770 : FAILURE in 29m 23s","accounts_in_message":[],"_revision_number":4},{"id":"21fd19f3d5513e4fcd7c5b3a4ee4c1af13cac1ce","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-05-11 18:39:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b07cb9797089e5d076175c53fe58c8f355ac721c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-05-13 20:47:58.000000000","message":"Patch Set 4: Code-Review+2\n\nI still think okay to allow ping in any namespace, so the convoluted regexp is weird.\n\nbut if it works fine by me","accounts_in_message":[],"_revision_number":4},{"id":"43dcde077433bcf6df24fe820dfd7de859026ad9","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-05-14 08:38:59.000000000","message":"Patch Set 4:\n\nI found no way to express \"a variable number of characters, but no space, meaning only one argument of\".","accounts_in_message":[],"_revision_number":4},{"id":"ec21965ad0b8ebce8c52522fba1252da39b65495","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-05-14 18:52:18.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"7c3189ed3202347e44ac4c135458a57e0b14b696","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-14 18:54:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/92273fd4568349de8a02cadeef93cefc : POST_FAILURE in 1m 54s","accounts_in_message":[],"_revision_number":5},{"id":"b0ee56294da09a6374038142908ee06e14957450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:00:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/362fbd671e68459f9bb09325f7593a3d : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":5},{"id":"01af43cfbe0b50415167b14954de224e72aa6734","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-05-14 19:50:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-c8ca10e6a9c6fcb6e5ba54646f1daf38ac962fe0 : FAILURE in 57m 27s","accounts_in_message":[],"_revision_number":5},{"id":"c8c3038f370dbd76caa5756d4b8822ae6842b6e0","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-05-15 10:27:50.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1e73559b6857cf752cdcea9e2477120bd437a77a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 10:28:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9b903ff9c5b2fede1653d9808b6152a2a5f9ffbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 10:34:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9319b3599bc3440599700f2f92c8c4c8 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":5},{"id":"cefab7e66a3abb6c163c4618f72e0fd2a981fb85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 10:34:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c8ca10e6a9c6fcb6e5ba54646f1daf38ac962fe0","revisions":{"00077266ae8d70e5d79eb65fdb8081150cbc7842":{"kind":"REWORK","_number":1,"created":"2020-05-08 14:44:36.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/06/726406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/06/726406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/1"}}},"commit":{"parents":[{"commit":"91b9f4c4dfca95174a1351f5b69efc2eea0f5a47","subject":"Merge \"metalsmith: update to 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/91b9f4c4dfca95174a1351f5b69efc2eea0f5a47"}]}],"author":{"name":"git","email":"keith.berger@suse.com","date":"2020-05-08 14:43:04.000000000","tz":-240},"committer":{"name":"git","email":"keith.berger@suse.com","date":"2020-05-08 14:43:04.000000000","tz":-240},"subject":"Lockdown /bin/ip permissions for the monasca-agent","message":"Lockdown /bin/ip permissions for the monasca-agent\n\nThis patch adds addtional arguments to the sudoers entry for\nthe /bin/ip command. It restricts access to only \u0027ip netns exec\u0027.\n\nChange-Id: Ie80c8fbdc851cbace8c82f8c47f490898f5c4d6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/00077266ae8d70e5d79eb65fdb8081150cbc7842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/00077266ae8d70e5d79eb65fdb8081150cbc7842"}]},"branch":"refs/heads/master"},"5f8058dc47d7125a73be8c12c9ef3b48c5be82ba":{"kind":"REWORK","_number":2,"created":"2020-05-11 15:40:30.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/06/726406/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/06/726406/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/2"}}},"commit":{"parents":[{"commit":"91b9f4c4dfca95174a1351f5b69efc2eea0f5a47","subject":"Merge \"metalsmith: update to 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/91b9f4c4dfca95174a1351f5b69efc2eea0f5a47"}]}],"author":{"name":"git","email":"keith.berger@suse.com","date":"2020-05-08 14:43:04.000000000","tz":-240},"committer":{"name":"KeithMnemonic","email":"keith.berger@suse.com","date":"2020-05-11 15:40:23.000000000","tz":-240},"subject":"Lockdown /bin/ip permissions for the monasca-agent","message":"Lockdown /bin/ip permissions for the monasca-agent\n\nThis patch adds addtional arguments to the sudoers entry for\nthe /bin/ip command. It restricts access to only \u0027ip netns exec\u0027.\n\nChange-Id: Ie80c8fbdc851cbace8c82f8c47f490898f5c4d6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5f8058dc47d7125a73be8c12c9ef3b48c5be82ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5f8058dc47d7125a73be8c12c9ef3b48c5be82ba"}]},"branch":"refs/heads/master"},"ce26bbbbb9afe35a2d6354bf9e225aa2e374afc6":{"kind":"REWORK","_number":3,"created":"2020-05-11 17:54:08.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/06/726406/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/06/726406/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/3"}}},"commit":{"parents":[{"commit":"91b9f4c4dfca95174a1351f5b69efc2eea0f5a47","subject":"Merge \"metalsmith: update to 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/91b9f4c4dfca95174a1351f5b69efc2eea0f5a47"}]}],"author":{"name":"git","email":"keith.berger@suse.com","date":"2020-05-08 14:43:04.000000000","tz":-240},"committer":{"name":"KeithMnemonic","email":"keith.berger@suse.com","date":"2020-05-11 17:53:55.000000000","tz":-240},"subject":"Lockdown /bin/ip permissions for the monasca-agent","message":"Lockdown /bin/ip permissions for the monasca-agent\n\nThis patch adds addtional arguments to the sudoers entry for\nthe /bin/ip command. It restricts access to only \u0027ip netns exec\u0027.\n\nChange-Id: Ie80c8fbdc851cbace8c82f8c47f490898f5c4d6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ce26bbbbb9afe35a2d6354bf9e225aa2e374afc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ce26bbbbb9afe35a2d6354bf9e225aa2e374afc6"}]},"branch":"refs/heads/master"},"6ff1f39f00c6ace3dcc528c2187cd5d93673fe87":{"kind":"REWORK","_number":4,"created":"2020-05-11 18:06:12.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/06/726406/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/06/726406/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/4"}}},"commit":{"parents":[{"commit":"91b9f4c4dfca95174a1351f5b69efc2eea0f5a47","subject":"Merge \"metalsmith: update to 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/91b9f4c4dfca95174a1351f5b69efc2eea0f5a47"}]}],"author":{"name":"git","email":"keith.berger@suse.com","date":"2020-05-08 14:43:04.000000000","tz":-240},"committer":{"name":"KeithMnemonic","email":"keith.berger@suse.com","date":"2020-05-11 18:05:58.000000000","tz":-240},"subject":"Lockdown /bin/ip permissions for the monasca-agent","message":"Lockdown /bin/ip permissions for the monasca-agent\n\nThis patch adds addtional arguments to the sudoers entry for\nthe /bin/ip command. It restricts access to only \u0027ip netns exec\u0027.\n\nChange-Id: Ie80c8fbdc851cbace8c82f8c47f490898f5c4d6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6ff1f39f00c6ace3dcc528c2187cd5d93673fe87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6ff1f39f00c6ace3dcc528c2187cd5d93673fe87"}]},"branch":"refs/heads/master"},"c8ca10e6a9c6fcb6e5ba54646f1daf38ac962fe0":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-05-14 18:52:18.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/06/726406/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/06/726406/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/06/726406/5"}}},"commit":{"parents":[{"commit":"f7a92f5b5849cdf8c7cef336bb532bd77e781a1c","subject":"Merge \"monasca-common: Add simplejson\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/f7a92f5b5849cdf8c7cef336bb532bd77e781a1c"}]}],"author":{"name":"git","email":"keith.berger@suse.com","date":"2020-05-08 14:43:04.000000000","tz":-240},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-05-14 18:52:18.000000000","tz":0},"subject":"Lockdown /bin/ip permissions for the monasca-agent","message":"Lockdown /bin/ip permissions for the monasca-agent\n\nThis patch adds addtional arguments to the sudoers entry for\nthe /bin/ip command. It restricts access to only \u0027ip netns exec\u0027.\n\nChange-Id: Ie80c8fbdc851cbace8c82f8c47f490898f5c4d6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/c8ca10e6a9c6fcb6e5ba54646f1daf38ac962fe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/c8ca10e6a9c6fcb6e5ba54646f1daf38ac962fe0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
