)]}'
{"id":"openstack%2Frpm-packaging~733018","triplet_id":"openstack%2Frpm-packaging~stable%2Fussuri~I38b8bac845b3b6a7c043edb849d257ea0fc90cb9","project":"openstack/rpm-packaging","branch":"stable/ussuri","hashtags":[],"change_id":"I38b8bac845b3b6a7c043edb849d257ea0fc90cb9","subject":"murano-pkg-check: Blacklist failing test","status":"MERGED","created":"2020-06-02 22:15:41.000000000","updated":"2020-06-04 10:43:16.000000000","submitted":"2020-06-04 10:43:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733018-1591267396674-0e537f8a","meta_rev_id":"a9f7cabd752ff7380afcc84551ca5c992aaff2a5","_number":733018,"virtual_id_number":733018,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2020-06-02 22:50:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 10:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-03 08:04:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-03 12:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-04 10:34:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 22:50:26.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-03 08:04:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-03 12:51:34.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-06-04 10:43:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a55c606cbc176591e290d4a8b181413be96fec4","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-06-02 22:15:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f1f9244ae95b917cedb5020e533da5d283e5af1c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-02 22:19:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/5853b3f32246440a8043ab78bcb9702e : FAILURE in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"144628b6dca9d69fc7887bc9b5b0848f86597e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 22:29:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d0f6f57deb1242e992cf2d19251f574f : FAILURE in 13m 12s","accounts_in_message":[],"_revision_number":1},{"id":"075df69556e8f3b99b0853fa3251844007732a7c","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-06-02 22:50:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Ussuri https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Ussuri-7603f3f4127b624988c09d26ca5bafe2c9a0f65a : FAILURE in 34m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8d863291f17b9ca6567aea9a3bf8aabcb0fe5277","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-03 07:56:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"844a6d5ac4231a25ab143622de468331c3b196c3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-03 08:04:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/d2ba8e0dd1c147e4a368d038b4035672 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"613603d81772aba82f19c011facca25b616d3378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 08:06:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5279111634f1471e833aa63999921cb6 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1e75ea151ba292ac2f21ae43fc1c6579bcf32d15","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-06-03 08:21:13.000000000","message":"Patch Set 1:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Ussuri https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Ussuri-7603f3f4127b624988c09d26ca5bafe2c9a0f65a : FAILURE in 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1ebde66b315115ca2b32dd08930549ea44335076","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-03 12:51:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5219367c88c0ac372ef6bba8b5671301f5d3dce","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-06-04 10:34:37.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d97d14f7897215952a51f3e612bbd690a27a55c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:34:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41ca4bb1d87973886b0b2b2209341a33196fe282","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:43:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9eb6087013c74f6ebfa9d86f5cacd3af : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a9f7cabd752ff7380afcc84551ca5c992aaff2a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:43:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7603f3f4127b624988c09d26ca5bafe2c9a0f65a","revisions":{"7603f3f4127b624988c09d26ca5bafe2c9a0f65a":{"kind":"REWORK","_number":1,"created":"2020-06-02 22:15:41.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/18/733018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/18/733018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/18/733018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/18/733018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/18/733018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/18/733018/1"}}},"commit":{"parents":[{"commit":"86c92f218b57969df86a5aa404d248c337a958ca","subject":"openstack-doc-tools: add obsoletes for python 2.x version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/86c92f218b57969df86a5aa404d248c337a958ca"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-06-02 11:42:16.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-06-02 22:15:41.000000000","tz":0},"subject":"murano-pkg-check: Blacklist failing test","message":"murano-pkg-check: Blacklist failing test\n\nThis also fails upstream atm\n\nChange-Id: I38b8bac845b3b6a7c043edb849d257ea0fc90cb9\n(cherry picked from commit 908e1ce7bd3bc79d08b3c07cb0d664eddd32cf24)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7603f3f4127b624988c09d26ca5bafe2c9a0f65a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7603f3f4127b624988c09d26ca5bafe2c9a0f65a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
