)]}'
{"id":"openstack%2Frpm-packaging~747811","triplet_id":"openstack%2Frpm-packaging~master~I7e5295f19829eacb299fdc421f8f36eff6451285","project":"openstack/rpm-packaging","branch":"master","topic":"bug/123456789","hashtags":[],"change_id":"I7e5295f19829eacb299fdc421f8f36eff6451285","subject":"Aug 25th updates","status":"ABANDONED","created":"2020-08-25 01:30:33.000000000","updated":"2020-09-16 18:25:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0b240e9b1a45ff572fbbf04b14dd0cbd4ebe9cba","_number":747811,"virtual_id_number":747811,"owner":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2020-08-25 02:08:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-25 01:59:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-08-25 01:56:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-26 09:52:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-28 20:34:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 01:56:17.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-25 01:59:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-25 02:08:20.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-26 09:52:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-28 20:34:54.000000000","updated_by":{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"},"reviewer":{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"},"state":"REVIEWER"}],"messages":[{"id":"d16cc42a6745ff90aceac9dc0b520b26853fc82d","author":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"date":"2020-08-25 01:30:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8db1cc9891ab1873bc10ed3eaa108ab729c8c387","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-25 01:36:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/de2cf0ad40c64f95b86c53e329a8f3e6 : FAILURE in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a24677afb4407a6b0d77fd9910a310c61483edb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 01:39:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ab9e7ae279f743cb869753cfbea7d9ba : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2745ca0280eda0cdc05f8b3e8fe44d2ceb4adc73","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-08-25 01:48:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-29ac77729cc20a8750ced84c83e740cdb169f7b9 : FAILURE in 17m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f8bc73403e933980652e311051e00bb555bf417c","author":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"date":"2020-08-25 01:50:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"44df626e91bebd1e0e0e4ea0ddcef3a678d64355","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-25 01:56:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/a83c246cbd46448aae7f98450d187255 : FAILURE in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"0d1161e79db6c82943f38ba2194b762a0b0f25f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 01:59:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fc4030ea747842c2906d4904eb6f0397 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"22d507b0e29e7810d3703cabd5109c9d63a6feaf","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-08-25 02:08:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-16cf940b97960e7c403361aa0f854a7607ef1ca3 : FAILURE in 17m 45s","accounts_in_message":[],"_revision_number":2},{"id":"82f0497bcbc0fe699a791622adeb708cc0ce41c4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-26 09:52:34.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM\n\nhttps://github.com/openstack/releases/commit/2d8bf89475a501f1f3bffde88790fab45c341604","accounts_in_message":[],"_revision_number":2},{"id":"eff5cc15652634464d4f2c640594f806748f3ef6","author":{"_account_id":17246,"name":"EB","email":"esampson@suse.com","username":"sampsone"},"date":"2020-08-28 20:34:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c0d3841b40b48cf0459c1159852cfc10320fe468","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-09-16 18:25:49.000000000","message":"Abandoned\n\nobsoleted by https://review.opendev.org/#/c/752288/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"16cf940b97960e7c403361aa0f854a7607ef1ca3","revisions":{"7f7dd522e9cec66748c2acea3f72bf1295f02755":{"kind":"REWORK","_number":1,"created":"2020-08-25 01:30:33.000000000","uploader":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"ref":"refs/changes/11/747811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/11/747811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/747811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/747811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/747811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/11/747811/1"}}},"commit":{"parents":[{"commit":"8b38bfd6643a783fb15459d56a1eb6b8443eb184","subject":"Merge \"update oslo.config to 8.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/8b38bfd6643a783fb15459d56a1eb6b8443eb184"}]}],"author":{"name":"zhangyc","email":"zhyanchao1988@126.com","date":"2020-08-25 01:30:05.000000000","tz":480},"committer":{"name":"zhangyc","email":"zhyanchao1988@126.com","date":"2020-08-25 01:30:05.000000000","tz":480},"subject":"Aug 25th updates","message":"Aug 25th updates\n\nChange-Id: I7e5295f19829eacb299fdc421f8f36eff6451285\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7f7dd522e9cec66748c2acea3f72bf1295f02755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/7f7dd522e9cec66748c2acea3f72bf1295f02755"}]},"branch":"refs/heads/master"},"16cf940b97960e7c403361aa0f854a7607ef1ca3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-25 01:50:16.000000000","uploader":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"ref":"refs/changes/11/747811/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/11/747811/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/747811/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/747811/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/747811/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/11/747811/2"}}},"commit":{"parents":[{"commit":"fcf62b490d4c87a4f5b6e4395cf2a4df28c9599c","subject":"Merge \"update dependencies to follow the package renaming on SUSE side\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/fcf62b490d4c87a4f5b6e4395cf2a4df28c9599c"}]}],"author":{"name":"zhangyc","email":"zhyanchao1988@126.com","date":"2020-08-25 01:30:05.000000000","tz":480},"committer":{"name":"zhangyc","email":"zhyanchao1988@126.com","date":"2020-08-25 01:50:16.000000000","tz":0},"subject":"Aug 25th updates","message":"Aug 25th updates\n\nChange-Id: I7e5295f19829eacb299fdc421f8f36eff6451285\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/16cf940b97960e7c403361aa0f854a7607ef1ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/16cf940b97960e7c403361aa0f854a7607ef1ca3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
