)]}'
{"id":"openstack%2Frpm-packaging~756011","triplet_id":"openstack%2Frpm-packaging~stable%2Ftrain~I82c8e625cc190570d7256f1f9648fcb6dbb4b9a3","project":"openstack/rpm-packaging","branch":"stable/train","topic":"train","hashtags":[],"change_id":"I82c8e625cc190570d7256f1f9648fcb6dbb4b9a3","subject":"Update oslo.privsep to 1.33.4.","status":"MERGED","created":"2020-10-05 08:41:52.000000000","updated":"2020-10-07 18:13:54.000000000","submitted":"2020-10-07 18:13:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756011-1602094434161-e81968aa","meta_rev_id":"ded507a1cf4667975d17b6e4232e238e58ef0349","_number":756011,"virtual_id_number":756011,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2020-10-07 10:56:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-07 18:13:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-07 09:04:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-07 18:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2020-10-07 14:53:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-07 09:00:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-07 18:01:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 09:00:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-07 09:04:50.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-07 10:56:41.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-07 14:53:40.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-07 18:01:44.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-10-07 18:13:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09c2e8d8f57d043f165db944f9c96f444c5b2bd9","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-05 08:41:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b6ad5a975ab8e7909491bd0165885dcaf0e9677","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 08:43:20.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"50b866baa95ecbca16c5a21e721333443ea124fd","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-10-05 08:44:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-fbfdb1d762b2a3187be0c7a0b02e88502672449a : FAILURE in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"024b37dea33d33e0e09a12288f51d4539a736edb","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-05 08:45:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"463ee0b8e86ccd00eb963f5f8efb4da1116bdcc6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-05 08:47:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/a0195446986f494bbeb7f39b87001f19 : FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4c03135fd6080cbe01d2b8623a53e9d7e8436eb0","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-10-05 08:47:37.000000000","message":"Patch Set 1:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-fbfdb1d762b2a3187be0c7a0b02e88502672449a : FAILURE in 1m 59s","accounts_in_message":[],"_revision_number":1},{"id":"98c99bd3fcf6d18d090b727f6ea432cf658e7a95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 08:52:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2be0ec6ba8cc45f7b42738301bc1be56 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9ce19a1b0eb63ff6371c96503df6272f1fb99557","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 09:57:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0d0e8116f7e6ebdef4b0afc13b01f4162536d9b1","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-05 09:57:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"888f2d5e09e839d08b3391e1616630107fd427f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-05 10:03:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/e94fa93cdd70400490a4f3d56073dd21 : FAILURE in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7f97f38c29dd03d0224f0bf0619bac538af739e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 10:10:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d330794ec9ce40e88d047b141021ab5a : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6b0034b7dd109c3cefe552aa7c4fd29dd25d519c","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-05 10:25:35.000000000","message":"Patch Set 1: Code-Review-1\n\nYou need to remove the patch :\n0001-Disable-logger-validation-during-unit-testing.patch\n\nIt is included in commit 00e69d4f32af14a49727b0b9dcad614d8dd29edb","accounts_in_message":[],"_revision_number":1},{"id":"70dfa8ab8bec8bc16ea478e6566d0f092b0d3b5a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 10:27:16.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e You need to remove the patch :\n \u003e 0001-Disable-logger-validation-during-unit-testing.patch\n \u003e \n \u003e It is included in commit 00e69d4f32af14a49727b0b9dcad614d8dd29edb\n\nAh yes I missed that one","accounts_in_message":[],"_revision_number":1},{"id":"85c7c8030921ae8fd68952d0cc339aebe676066e","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-10-05 10:44:39.000000000","message":"Patch Set 1:\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-fbfdb1d762b2a3187be0c7a0b02e88502672449a : FAILURE in 47m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8741b1fb5d2675d2e2590a75ca36ac6fd4704364","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-05 13:08:53.000000000","message":"Patch Set 1:\n\nBy the way, why do you need to increase version of oslo.privsep on a stable branch ?\n\nIs it related to bug resolution ?","accounts_in_message":[],"_revision_number":1},{"id":"38c9629405c5b247cc536e0dd21107e0433f0910","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 13:48:37.000000000","message":"Patch Set 1:\n\n\u003e By the way, why do you need to increase version of oslo.privsep on\n \u003e a stable branch ?\n \u003e \n \u003e Is it related to bug resolution ?\n\nYep we released a minor version to land an important fix","accounts_in_message":[],"_revision_number":1},{"id":"4bfbd7d79ea7621d61ab783050763e3cdea5e451","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-05 13:52:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a16b1892b59cc483c788edf856062482b4ad104","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 13:53:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57487591195b3157f73eff6174942dd9ca58b18d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-05 13:57:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/aa3b67ef189146a0bd3ea57d195958e9 : FAILURE in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"01f82abbc228cc6bb9702d1501dfd6361f2b76aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 14:05:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2804805c9f434994aab68aab8c8c8092 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"b09076c6b2df52dbf428abd8ef6d5ae031a6d7d3","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-10-05 15:01:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-9dd531b09eac12f30d7637b526068b23dc96e9d3 : FAILURE in 1h 09m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ede8ef9fd3f7518b630d2f8e3abb8e6b62acbc32","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-07 08:57:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"22c4a494ae8a3d7ebb307390db92a7b88676942c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-07 09:00:05.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"fd2b7d7075e29b92bd4dcd2525439f16915edf52","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-07 09:04:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/27b656bfcd9f41f09f731a037ca44abe : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":3},{"id":"a60fe7e0e8fa3ce3fa16744c5fcbca7ec12954d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 09:07:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/18ae3ecf8a024831a47b782bcb27ba4b : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"cecc1b366c11c0d24346067b9f580dc4e19787c8","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2020-10-07 10:56:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Train https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Train-a2fb51fcc8a7158442ba8ff1183cfc4b16b0ef83 : SUCCESS in 1h 59m 08s","accounts_in_message":[],"_revision_number":3},{"id":"500025bb950d59bb9620656009eecf223ffc43ae","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-07 14:53:40.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f50b5d057fb2f49a8c4835817f57fcb4e8d961bf","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-10-07 18:01:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2d177cf11140210b0382d23e653be52a8391967a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:01:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d949e7d65f74781322525d6f50cb8c8a194cd163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:13:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6db8e08c8b90414483302fdf659d14f4 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"72dfc9fc7d78a8938b4c84db4c4aa5699d3d3c09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:13:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a2fb51fcc8a7158442ba8ff1183cfc4b16b0ef83","revisions":{"fbfdb1d762b2a3187be0c7a0b02e88502672449a":{"kind":"REWORK","_number":1,"created":"2020-10-05 08:41:52.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/11/756011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/11/756011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/1"}}},"commit":{"parents":[{"commit":"a82162e1c05e5deca3ce08936382a50d60ce37b5","subject":"Train Updates (july 2020)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a82162e1c05e5deca3ce08936382a50d60ce37b5"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-10-05 08:47:35.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-10-05 08:47:35.000000000","tz":120},"subject":"Update oslo.privsep to 1.33.4.","message":"Update oslo.privsep to 1.33.4.\n\nUpdate the version to the new release.\n\nChange-Id: I82c8e625cc190570d7256f1f9648fcb6dbb4b9a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/fbfdb1d762b2a3187be0c7a0b02e88502672449a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/fbfdb1d762b2a3187be0c7a0b02e88502672449a"}]},"branch":"refs/heads/stable/train"},"9dd531b09eac12f30d7637b526068b23dc96e9d3":{"kind":"REWORK","_number":2,"created":"2020-10-05 13:52:02.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/11/756011/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/11/756011/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/2"}}},"commit":{"parents":[{"commit":"a82162e1c05e5deca3ce08936382a50d60ce37b5","subject":"Train Updates (july 2020)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a82162e1c05e5deca3ce08936382a50d60ce37b5"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-10-05 08:47:35.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-10-05 13:57:52.000000000","tz":120},"subject":"Update oslo.privsep to 1.33.4.","message":"Update oslo.privsep to 1.33.4.\n\nUpdate the version to the new release.\n\nChange-Id: I82c8e625cc190570d7256f1f9648fcb6dbb4b9a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9dd531b09eac12f30d7637b526068b23dc96e9d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9dd531b09eac12f30d7637b526068b23dc96e9d3"}]},"branch":"refs/heads/stable/train"},"a2fb51fcc8a7158442ba8ff1183cfc4b16b0ef83":{"kind":"REWORK","_number":3,"created":"2020-10-07 08:57:12.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/11/756011/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/11/756011/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/11/756011/3"}}},"commit":{"parents":[{"commit":"a82162e1c05e5deca3ce08936382a50d60ce37b5","subject":"Train Updates (july 2020)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a82162e1c05e5deca3ce08936382a50d60ce37b5"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-10-05 08:47:35.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-10-07 09:03:02.000000000","tz":120},"subject":"Update oslo.privsep to 1.33.4.","message":"Update oslo.privsep to 1.33.4.\n\nUpdate the version to the new release.\n\nChange-Id: I82c8e625cc190570d7256f1f9648fcb6dbb4b9a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a2fb51fcc8a7158442ba8ff1183cfc4b16b0ef83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/a2fb51fcc8a7158442ba8ff1183cfc4b16b0ef83"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
