)]}'
{"id":"openstack%2Frpm-packaging~775967","triplet_id":"openstack%2Frpm-packaging~master~I05f8c8458f13ab18e68b248689a513682ef50f38","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"I05f8c8458f13ab18e68b248689a513682ef50f38","subject":"cinder: add boto3 dependency","status":"MERGED","created":"2021-02-17 07:48:28.000000000","updated":"2021-03-12 14:09:55.000000000","submitted":"2021-03-12 14:09:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"775967","meta_rev_id":"13819e717b9ef7ef1f64b48356102d9029b3b3ec","_number":775967,"virtual_id_number":775967,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2021-03-11 18:35:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 14:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-11 17:43:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-11 17:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-12 14:02:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 07:56:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-17 08:01:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-17 08:29:09.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-18 13:51:38.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"CC"},{"updated":"2021-03-11 17:43:23.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"}],"messages":[{"id":"6192c4498148356e87ba9ef87c0ffc66243a21cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2021-02-17 07:48:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec148526ddbc6fd0f89b014365d9b32229ba4ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 07:56:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bbba1973560b4fa8a3877aedfca5d645 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8d3ab50396dd1a02252c93358ef0233a83d33ae5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-17 08:01:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/8ecd0d77cf6c4899af6d13205737c101 : FAILURE in 11m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ba8a63fbb09c27ae5c3ff4b7125d625912bafabb","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2021-02-17 08:29:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-0de2f11b08cfdabc7c7a94a1f8a0284e2d63aa5c : FAILURE in 40m 20s","accounts_in_message":[],"_revision_number":1},{"id":"670e7e2e462b4d2fe345d5a1bb0b1b1c92ddb535","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-02-18 13:51:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6632968c88c0ebca29590abf8323b9706af66c19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2021-02-23 21:30:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ebd1a9fd48f9d1d48a631a79932cc4595be968d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 21:37:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ffbfe6ad9fbf4759b5ea5bf9b7ba1cf7 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"cb8ba79fdd5a6e3d74aec4b6a1bd3e68c76439b2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-23 21:41:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/dc7fc03aed2546a8b5a4abaac1dbe69c : FAILURE in 11m 13s","accounts_in_message":[],"_revision_number":2},{"id":"abd66ad23dcd6e199d70dbbc315757da1b7fa62e","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2021-02-23 21:45:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-ef0500b26c93f6b5aa9c784db94b897cf4fe64b1 : FAILURE in 15m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4133aa17c2b628d766de1038649d232f473842b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-02-24 09:29:53.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1507ea0c84af2a9b798f816b4348f3b3ed1f72de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 09:41:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5aef4b20579447df8006f8645ba94400 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1e40a4753b5d8b4b7d4f4272b5ff48d0cf12f9b9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-24 09:46:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/21399be2ce7e4f4092a32b774ac5409e : SUCCESS in 15m 42s","accounts_in_message":[],"_revision_number":3},{"id":"19cf3bb28157f8d90c722b249679c767f1fdd70b","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2021-02-24 12:08:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-5de630d34512c1343e3ae9d94a76d73c0693356e : FAILURE in 2h 37m 37s","accounts_in_message":[],"_revision_number":3},{"id":"9340ac51dfe95f7fc48bfda52bbbc1fd421fe5c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2021-03-11 17:07:07.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"99ecd27b8248d39723cca4e87ddf7aadd2701fa2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-11 17:17:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/25f708bd75ab490a9d01436b6b437532 : FAILURE in 9m 39s","accounts_in_message":[],"_revision_number":4},{"id":"2898ba463bd297dfb045c1b637bcce28a0691aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 17:18:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dc87b7d72a7444149da1331a0707f858 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":4},{"id":"5abc42469533fd4f9c67264ca8ab56a48c977992","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-03-11 17:26:13.000000000","message":"Patch Set 4:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":4},{"id":"31679714ac1084b80cdb0a2ff300aff98a3c5e48","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-11 17:43:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/4d702928704f47a99f1c30f23e6a49bb : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":4},{"id":"b542ac25e2ca73c8bcbb8c3380c3f36469a0f8d8","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-03-11 17:43:23.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"15de7ca8180dcdb975cbb42dd82c6dd5e9d3aeda","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2021-03-11 18:35:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-d1764e42cd9079fc0e1be4dc90fe475fcd42f035 : FAILURE in 1h 27m 38s","accounts_in_message":[],"_revision_number":4},{"id":"26247c40f5a1fcd5cb30dc49be1fc3154ed91b3f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2021-03-12 14:02:45.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"31c9a8298c85a774fff9e7ddd558e6d2571f2adf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 14:02:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7f381c6f0daa04732870db0e88e1264c30ecf2eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 14:09:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ea3737b4da9d4b1b9f009a686a604f7d : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":4},{"id":"13819e717b9ef7ef1f64b48356102d9029b3b3ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 14:09:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d1764e42cd9079fc0e1be4dc90fe475fcd42f035","revisions":{"0de2f11b08cfdabc7c7a94a1f8a0284e2d63aa5c":{"kind":"REWORK","_number":1,"created":"2021-02-17 07:48:28.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/67/775967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/67/775967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/1"}}},"commit":{"parents":[{"commit":"3a09516b12747a20d849a1124150643fd8ddf041","subject":"Merge \"Update versions of ironic related projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3a09516b12747a20d849a1124150643fd8ddf041"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2021-02-17 07:47:47.000000000","tz":60},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2021-02-17 07:48:20.000000000","tz":60},"subject":"cinder: add boto3 dependency","message":"cinder: add boto3 dependency\n\ncinder in master requires boto3 now, otherwise you get:\n\n   File \"/home/abuild/rpmbuild/BUILD/cinder-17.1.0.dev218/cinder/backup/drivers/s3.py\", line 74, in \u003cmodule\u003e\n     import boto3\n   ModuleNotFoundError: No module named \u0027boto3\u0027\n\nChange-Id: I05f8c8458f13ab18e68b248689a513682ef50f38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0de2f11b08cfdabc7c7a94a1f8a0284e2d63aa5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/0de2f11b08cfdabc7c7a94a1f8a0284e2d63aa5c"}]},"branch":"refs/heads/master"},"ef0500b26c93f6b5aa9c784db94b897cf4fe64b1":{"kind":"REWORK","_number":2,"created":"2021-02-23 21:30:08.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/67/775967/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/67/775967/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/2"}}},"commit":{"parents":[{"commit":"3a09516b12747a20d849a1124150643fd8ddf041","subject":"Merge \"Update versions of ironic related projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3a09516b12747a20d849a1124150643fd8ddf041"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2021-02-17 07:47:47.000000000","tz":60},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2021-02-23 21:04:50.000000000","tz":60},"subject":"cinder: add boto3 dependency","message":"cinder: add boto3 dependency\n\ncinder in master requires boto3 now, otherwise you get:\n\n   File \"/home/abuild/rpmbuild/BUILD/cinder-17.1.0.dev218/cinder/backup/drivers/s3.py\", line 74, in \u003cmodule\u003e\n     import boto3\n   ModuleNotFoundError: No module named \u0027boto3\u0027\n\nChange-Id: I05f8c8458f13ab18e68b248689a513682ef50f38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ef0500b26c93f6b5aa9c784db94b897cf4fe64b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/ef0500b26c93f6b5aa9c784db94b897cf4fe64b1"}]},"branch":"refs/heads/master"},"5de630d34512c1343e3ae9d94a76d73c0693356e":{"kind":"REWORK","_number":3,"created":"2021-02-24 09:29:53.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/67/775967/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/67/775967/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/3"}}},"commit":{"parents":[{"commit":"3a09516b12747a20d849a1124150643fd8ddf041","subject":"Merge \"Update versions of ironic related projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/3a09516b12747a20d849a1124150643fd8ddf041"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2021-02-17 07:47:47.000000000","tz":60},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2021-02-24 09:29:48.000000000","tz":0},"subject":"cinder: add boto3 dependency","message":"cinder: add boto3 dependency\n\ncinder in master requires boto3 now, otherwise you get:\n\n   File \"/home/abuild/rpmbuild/BUILD/cinder-17.1.0.dev218/cinder/backup/drivers/s3.py\", line 74, in \u003cmodule\u003e\n     import boto3\n   ModuleNotFoundError: No module named \u0027boto3\u0027\n\nChange-Id: I05f8c8458f13ab18e68b248689a513682ef50f38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5de630d34512c1343e3ae9d94a76d73c0693356e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5de630d34512c1343e3ae9d94a76d73c0693356e"}]},"branch":"refs/heads/master"},"d1764e42cd9079fc0e1be4dc90fe475fcd42f035":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-11 17:07:07.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/67/775967/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/67/775967/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/67/775967/4"}}},"commit":{"parents":[{"commit":"c6dfef9b130287016137564269e08e84879c0438","subject":"Merge \"Wallaby updates oslo packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/c6dfef9b130287016137564269e08e84879c0438"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2021-02-17 07:47:47.000000000","tz":60},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2021-03-11 17:07:07.000000000","tz":0},"subject":"cinder: add boto3 dependency","message":"cinder: add boto3 dependency\n\ncinder in master requires boto3 now, otherwise you get:\n\n   File \"/home/abuild/rpmbuild/BUILD/cinder-17.1.0.dev218/cinder/backup/drivers/s3.py\", line 74, in \u003cmodule\u003e\n     import boto3\n   ModuleNotFoundError: No module named \u0027boto3\u0027\n\nChange-Id: I05f8c8458f13ab18e68b248689a513682ef50f38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d1764e42cd9079fc0e1be4dc90fe475fcd42f035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/d1764e42cd9079fc0e1be4dc90fe475fcd42f035"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
