)]}'
{"id":"openstack%2Frpm-packaging~787592","triplet_id":"openstack%2Frpm-packaging~master~Id4c920e300ce933c15e33932cff44eb5a37e0e4d","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"Id4c920e300ce933c15e33932cff44eb5a37e0e4d","subject":"nova: add nova-policy man page","status":"MERGED","created":"2021-04-22 15:53:38.000000000","updated":"2021-04-22 19:57:15.000000000","submitted":"2021-04-22 19:57:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787592","meta_rev_id":"dee5368a74fbe9aa49de1ac44f71b3d4eb860dac","_number":787592,"virtual_id_number":787592,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-04-22 17:00:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-22 19:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-22 16:16:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-22 19:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-22 19:29:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 16:02:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 16:16:41.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 17:00:01.000000000","updated_by":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 19:29:45.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"}],"messages":[{"id":"67d9359681a6cdd50e1fdd9016b6ede42445d175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-04-22 15:53:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f85b450cf1a24fe6e34ad5711f7a8f31a1d5964","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-04-22 15:54:57.000000000","message":"Patch Set 1:\n\nI see the Nova patch is only in master, so maybe we should merge https://review.opendev.org/c/openstack/rpm-packaging/+/787360 first, then this one, so we can cut stable/wallaby correctly.","accounts_in_message":[],"_revision_number":1},{"id":"df7a72cfc63136252d3be27b9500cc9a058d47f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 16:02:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0eaaf371b47c49cda2f27a376d902edc : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"01e6487115c76d61c72f0f0b21d5a86c4f46d528","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-22 16:16:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- DLRN-rpmbuild-rpm-packaging-centos-8 https://review.rdoproject.org/zuul/build/bb2ecb3f8d844fd1a94b7c7a7e8de3d6 : SUCCESS in 22m 26s","accounts_in_message":[],"_revision_number":1},{"id":"239b795b3acaf797a6b102ad6c82b63832de4457","author":{"_account_id":19648,"name":"SUSE Cloud CI","email":"dmueller+suse-cloud-ci@suse.de","username":"suse-ci","tags":["SERVICE_USER"]},"date":"2021-04-22 17:00:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (rpm-packaging).\n\n- openstack-rpm-packaging-sles15-Master https://build.opensuse.org/project/show/home:suse-cloud-ci:rpm-packaging-sles15-Master-10b98b4315cde5ce2055a409229b3b50abf8ea6d : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c59ba50a1f6fb8218f21b94e64281139d1031cab","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2021-04-22 19:29:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5293c1bed631a999bf9ab223d0b22c13bb52f7a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 19:29:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3fab5e06397206bf7b8c14e78180fc816ead276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 19:57:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/56a5b5d0fbd344a4991399228ef82c9f : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"dee5368a74fbe9aa49de1ac44f71b3d4eb860dac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 19:57:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10b98b4315cde5ce2055a409229b3b50abf8ea6d","revisions":{"10b98b4315cde5ce2055a409229b3b50abf8ea6d":{"kind":"REWORK","_number":1,"created":"2021-04-22 15:53:38.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/92/787592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/92/787592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/92/787592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/92/787592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/92/787592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/92/787592/1"}}},"commit":{"parents":[{"commit":"9eb78ae66693e46a06d540680732eaafcbed83dc","subject":"Merge \"Update to latest upper-constraints for os-*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/9eb78ae66693e46a06d540680732eaafcbed83dc"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-04-22 15:52:51.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-04-22 15:52:51.000000000","tz":120},"subject":"nova: add nova-policy man page","message":"nova: add nova-policy man page\n\nThis was added by https://review.opendev.org/c/openstack/nova/+/729918\n\nChange-Id: Id4c920e300ce933c15e33932cff44eb5a37e0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/10b98b4315cde5ce2055a409229b3b50abf8ea6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/10b98b4315cde5ce2055a409229b3b50abf8ea6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
