)]}'
{"id":"openstack%2Frpm-packaging~921631","triplet_id":"openstack%2Frpm-packaging~master~Iae81b36a1474b3c3219ae74bce7ad542fa4e60b8","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"Iae81b36a1474b3c3219ae74bce7ad542fa4e60b8","subject":"neutron-lib: add ddt dependency","status":"MERGED","created":"2024-06-09 21:01:00.000000000","updated":"2024-06-09 21:25:15.000000000","submitted":"2024-06-09 21:25:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921631","meta_rev_id":"a4c775561b88aaf8a01c8c9be411a47014654e77","_number":921631,"virtual_id_number":921631,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-09 21:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":2,"date":"2024-06-09 21:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":1,"date":"2024-06-09 21:06:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-09 21:10:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfb13d44ad95e493148dc83e5a866f931a611ad8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2024-06-09 21:01:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f178afda6882c103c735d58bd1d6d8151bee7a9","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2024-06-09 21:06:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6567c19973ca4cada37273643a8a6a7ff672c9f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 21:10:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acf9f4f088ef4bac9f0b93cd9ddb35fd\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/21a3153ce8dc42bf8985b648945f6d75 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8bad5b88c9a7d58ae5a83275765d6ff38b9af429","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 21:10:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7d6a8da2e07b34e87d03025b0c4c8fae35e8153","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 21:25:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8320ad1d80f04194a9814c799d0ed720\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b03747dcd3d744088c9b4283fc376f90 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a4c775561b88aaf8a01c8c9be411a47014654e77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 21:25:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1626cc72ab4510e4895a5ccb97e2514dd582506d","revisions":{"1626cc72ab4510e4895a5ccb97e2514dd582506d":{"kind":"REWORK","_number":1,"created":"2024-06-09 21:01:00.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/31/921631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/31/921631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/921631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/921631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/31/921631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/31/921631/1"}}},"commit":{"parents":[{"commit":"de3450198febd7147d9c8d23443a8148139a5ff9","subject":"Merge \"Fix typo in vitrageclient version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/de3450198febd7147d9c8d23443a8148139a5ff9"}]}],"author":{"name":"Dirk Müller","email":"dirk@dmllr.de","date":"2024-06-09 21:00:50.000000000","tz":120},"committer":{"name":"Dirk Müller","email":"dirk@dmllr.de","date":"2024-06-09 21:00:50.000000000","tz":120},"subject":"neutron-lib: add ddt dependency","message":"neutron-lib: add ddt dependency\n\nChange-Id: Iae81b36a1474b3c3219ae74bce7ad542fa4e60b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1626cc72ab4510e4895a5ccb97e2514dd582506d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/1626cc72ab4510e4895a5ccb97e2514dd582506d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
