)]}'
{"id":"openstack%2Frpm-packaging~942562","triplet_id":"openstack%2Frpm-packaging~master~Ib4e77ba3d934b6ac2d7a42fb2a837d309ef2e401","project":"openstack/rpm-packaging","branch":"master","hashtags":[],"change_id":"Ib4e77ba3d934b6ac2d7a42fb2a837d309ef2e401","subject":"pyghmi: update to 1.5.76","status":"MERGED","created":"2025-02-24 10:09:05.000000000","updated":"2025-02-24 10:47:55.000000000","submitted":"2025-02-24 10:47:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942562","meta_rev_id":"fff2ba0ce1cacb06bbaa9e1cfb2c251ac9a2fc1c","_number":942562,"virtual_id_number":942562,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-24 10:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":2,"date":"2025-02-24 10:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":1,"date":"2025-02-24 10:43:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-24 10:13:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4342940081963ba16241d72f5127f4fb7a8ec7e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2025-02-24 10:09:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3aecfdffc33ddd792927f8a498a4aaae3168251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 10:13:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e50d0e459843468de20bbfd6a8f309\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/328e299eccc44d789a8b9168c0772ea3 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9bd0d7d9cab51341b58da689f49472ee844cf5f2","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2025-02-24 10:43:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e6a89508f0dc36778b785dd4128943a1cac82ef6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 10:43:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d7684033b83215dab14f99462eddd8669266c35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 10:47:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6acb751a03374f87afaf97f34625ae94\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/578e5a6058114a3d9a13d970a085b80b : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fff2ba0ce1cacb06bbaa9e1cfb2c251ac9a2fc1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 10:47:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d4fe76d9b34e2c6a72dd3c22877b531562b51e6","revisions":{"5d4fe76d9b34e2c6a72dd3c22877b531562b51e6":{"kind":"REWORK","_number":1,"created":"2025-02-24 10:09:05.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/62/942562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rpm-packaging","ref":"refs/changes/62/942562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/62/942562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/62/942562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rpm-packaging refs/changes/62/942562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rpm-packaging refs/changes/62/942562/1"}}},"commit":{"parents":[{"commit":"6e8686f0cd5369e2541ca4ed08111a57ae64efc3","subject":"Remove retired projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/6e8686f0cd5369e2541ca4ed08111a57ae64efc3"}]}],"author":{"name":"Dirk Müller","email":"dirk@dmllr.de","date":"2025-02-24 10:07:51.000000000","tz":60},"committer":{"name":"Dirk Müller","email":"dirk@dmllr.de","date":"2025-02-24 10:07:51.000000000","tz":60},"subject":"pyghmi: update to 1.5.76","message":"pyghmi: update to 1.5.76\n\nChange-Id: Ib4e77ba3d934b6ac2d7a42fb2a837d309ef2e401\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5d4fe76d9b34e2c6a72dd3c22877b531562b51e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rpm-packaging/commit/5d4fe76d9b34e2c6a72dd3c22877b531562b51e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
