)]}'
{"id":"openstack%2Fsahara-dashboard~716421","triplet_id":"openstack%2Fsahara-dashboard~stable%2Ftrain~Ia2415a21f643648ebd02644224ec44dcc773713a","project":"openstack/sahara-dashboard","branch":"stable/train","hashtags":[],"change_id":"Ia2415a21f643648ebd02644224ec44dcc773713a","subject":"Fix pyScss version in lower-constraints.txt","status":"MERGED","created":"2020-03-31 20:19:09.000000000","updated":"2020-04-07 15:38:42.000000000","submitted":"2020-04-07 15:38:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716421-1586273922652-4a059950","meta_rev_id":"2bacc61c2325f146ef7478f2d33e05434595521b","_number":716421,"virtual_id_number":716421,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 15:38:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2020-04-07 15:25:33.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":2,"date":"2020-03-31 20:23:57.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2020-04-07 15:25:33.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"date":"2020-04-07 15:14:20.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 20:23:57.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-04-07 15:14:20.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2020-04-07 15:25:33.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2020-04-07 15:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19d042b2e2757963ea4aae39f2edd283377b4a3e","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-03-31 20:19:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6552707bbb118ad51101c67d6834fe433b392e6c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-31 20:23:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e5849c2d6b6762474c9c040e716bfc637f422cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 20:31:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/33ff3e45b3b8479dbfa03ff212dd5915 : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e44ed555a3f4ae58f5818a0167d6cbb : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88cf8b4adf8044dba86bcf96b32ab2cb : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8edde96571ba49768dc66b3b142c0d98 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96301ffd91f847c9b914b021104df7e1 : SUCCESS in 7m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b2a236949564365acb82ae4371d6457 : SUCCESS in 5m 38s\n- sahara-dashboard-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/f3e80caa9f584dbf81a8eb4f0453b7e0 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d232117d07c38ab3ae7c3b37a76a472a4759dd32","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2020-04-07 15:25:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aee0dc2f0777578dfa81c0482576bb5dde053810","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 15:25:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c87e8db6613059e3bd55339a541b0f52a0e2558f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 15:38:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/90686b3ec71343b08103a689537e6df3 : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f4eaf61aba34013a1b72ac73f3645f8 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caf33b5cfe2248648311f4f1888131d3 : SUCCESS in 7m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ca23ea02fdde402cb8434df87fa58ab3 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c07543dfa22b4bcfaaa16179c6e1e9eb : SUCCESS in 9m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64c15480c79f4f0f84e2d00cb8ba9c72 : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":1},{"id":"42edbb59a0a7682182529613e31fa7375661be66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 15:38:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f10355b08a0a66c999cde16623ce86409ec97ff1","revisions":{"f10355b08a0a66c999cde16623ce86409ec97ff1":{"kind":"REWORK","_number":1,"created":"2020-03-31 20:19:09.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/21/716421/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-dashboard","ref":"refs/changes/21/716421/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-dashboard refs/changes/21/716421/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-dashboard refs/changes/21/716421/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-dashboard refs/changes/21/716421/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-dashboard refs/changes/21/716421/1"}}},"commit":{"parents":[{"commit":"8846cb2157b04c559b133728f86d497380504238","subject":"Fix gate.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-dashboard/commit/8846cb2157b04c559b133728f86d497380504238"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-03-31 16:41:49.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2020-03-31 20:19:09.000000000","tz":0},"subject":"Fix pyScss version in lower-constraints.txt","message":"Fix pyScss version in lower-constraints.txt\n\nHorizon requires  pyScss\u003e\u003d1.3.7 [1] but lower-constraints.txt\nin sahara-dashboard uses pyScss 1.3.4 which can cause the\nfailure in future. So this patch update pyScss version in\nlower-constraints.txt to sync the change.\n\n[1] https://review.opendev.org/#/c/715241/\n\nChange-Id: Ia2415a21f643648ebd02644224ec44dcc773713a\n(cherry picked from commit 33d6f261c7472e7fee8502497f773bf1c6006881)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-dashboard/commit/f10355b08a0a66c999cde16623ce86409ec97ff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-dashboard/commit/f10355b08a0a66c999cde16623ce86409ec97ff1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
