)]}'
{"id":"openstack%2Fsahara-plugin-cdh~653993","triplet_id":"openstack%2Fsahara-plugin-cdh~master~I03ca82bcdd92f277ab7f34fe24ac6b5049e4bf95","project":"openstack/sahara-plugin-cdh","branch":"master","topic":"opendev","hashtags":[],"change_id":"I03ca82bcdd92f277ab7f34fe24ac6b5049e4bf95","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-20 13:00:13.000000000","updated":"2019-04-22 11:30:57.000000000","submitted":"2019-04-22 11:30:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653993-1555932657053-11ac1cc1","meta_rev_id":"ad2b6a943fce8b01be1462fd0b6d7267306af7fe","_number":653993,"virtual_id_number":653993,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2019-04-22 11:30:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2019-04-22 10:52:57.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2019-04-22 10:52:57.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 10:52:57.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-04-22 11:30:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7278bf27ae8215e359a9c86f9a5fb265580b123","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-20 13:00:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2577abd98e378d583176d880296089c53cd1bf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 13:24:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/93/653993/1/check/test-release-openstack/0c58b51/ : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/653993/1/check/openstack-tox-lower-constraints/8a00444/ : SUCCESS in 6m 24s\n- openstack-tox-pep8 http://logs.openstack.org/93/653993/1/check/openstack-tox-pep8/83eb451/ : SUCCESS in 6m 16s\n- openstack-tox-py27 http://logs.openstack.org/93/653993/1/check/openstack-tox-py27/09a4c00/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/93/653993/1/check/openstack-tox-py36/eeecedd/ : SUCCESS in 6m 15s\n- openstack-tox-py37 http://logs.openstack.org/93/653993/1/check/openstack-tox-py37/4329e38/ : SUCCESS in 4m 22s\n- openstack-tox-docs http://logs.openstack.org/93/653993/1/check/openstack-tox-docs/f7a0fab/html/ : SUCCESS in 7m 52s\n- build-openstack-releasenotes http://logs.openstack.org/93/653993/1/check/build-openstack-releasenotes/04f7f7d/html/ : SUCCESS in 5m 50s\n- sahara-buildimages-cloudera http://logs.openstack.org/93/653993/1/check/sahara-buildimages-cloudera/974e173/ : FAILURE in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8828dfe6059a10338204b9f66b155edabad5ddf","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-04-22 10:52:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5a193647cae109fb5f8ccb4f597e4736b51de75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 10:53:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d0dfcf4b7a084821fe964ea00d79b20d8deeb7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:30:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/93/653993/1/gate/test-release-openstack/05249be/ : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/653993/1/gate/openstack-tox-lower-constraints/d18a72e/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/93/653993/1/gate/openstack-tox-pep8/25577fe/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/93/653993/1/gate/openstack-tox-py27/41ebaca/ : SUCCESS in 7m 40s\n- openstack-tox-py36 http://logs.openstack.org/93/653993/1/gate/openstack-tox-py36/6447180/ : SUCCESS in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/93/653993/1/gate/openstack-tox-py37/4bcb532/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/93/653993/1/gate/openstack-tox-docs/e150f57/html/ : SUCCESS in 5m 14s\n- build-openstack-releasenotes http://logs.openstack.org/93/653993/1/gate/build-openstack-releasenotes/e31db8c/html/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3547f1c673201851696e31f043b5edf4565f04fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:30:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8fa3de4a508535085fcabc8872b38894619dfa98","revisions":{"8fa3de4a508535085fcabc8872b38894619dfa98":{"kind":"REWORK","_number":1,"created":"2019-04-20 13:00:13.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/93/653993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/93/653993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/93/653993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/93/653993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/93/653993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/93/653993/1"}}},"commit":{"parents":[{"commit":"66688aa1c36aed1ee9cff68ca54afe767d30fceb","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/66688aa1c36aed1ee9cff68ca54afe767d30fceb"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-18 13:12:12.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-20 11:51:41.000000000","tz":120},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org\ncounterparts.\n\nChange-Id: I03ca82bcdd92f277ab7f34fe24ac6b5049e4bf95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/8fa3de4a508535085fcabc8872b38894619dfa98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/8fa3de4a508535085fcabc8872b38894619dfa98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
