)]}'
{"id":"openstack%2Fsahara-plugin-cdh~654561","triplet_id":"openstack%2Fsahara-plugin-cdh~stable%2Fstein~I03ca82bcdd92f277ab7f34fe24ac6b5049e4bf95","project":"openstack/sahara-plugin-cdh","branch":"stable/stein","hashtags":[],"change_id":"I03ca82bcdd92f277ab7f34fe24ac6b5049e4bf95","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"ABANDONED","created":"2019-04-22 16:39:39.000000000","updated":"2019-04-25 14:40:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"80083260b52ef65184524c0ee6c5a5e2cc87250d","_number":654561,"virtual_id_number":654561,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-04-22 16:51:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 16:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ad2fbcab88498202f6e009d39b01b5629a5de85","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-22 16:39:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a1e4a780a07e6a876e051fe86b1e3bf4f9e977f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:51:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/61/654561/1/check/test-release-openstack/93cadf6/ : SUCCESS in 3m 00s\n- openstack-tox-pep8 http://logs.openstack.org/61/654561/1/check/openstack-tox-pep8/bda40ee/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/61/654561/1/check/openstack-tox-docs/3e6efb1/html/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"80083260b52ef65184524c0ee6c5a5e2cc87250d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-25 14:40:34.000000000","message":"Abandoned\n\nDuring today\u0027s Sahara meeting, and following -infra suggestion, it was deemed not necessary to backport those replacements on the older branches.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d69b8e9afc14b0c83cf066dd3b1ccd26526460c1","revisions":{"d69b8e9afc14b0c83cf066dd3b1ccd26526460c1":{"kind":"REWORK","_number":1,"created":"2019-04-22 16:39:39.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/61/654561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/61/654561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/61/654561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/61/654561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/61/654561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/61/654561/1"}}},"commit":{"parents":[{"commit":"b73a04d0c8c14686e2ade370406609277032bcb7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/b73a04d0c8c14686e2ade370406609277032bcb7"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-18 13:12:12.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-22 16:38:39.000000000","tz":120},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org\ncounterparts.\n\nChange-Id: I03ca82bcdd92f277ab7f34fe24ac6b5049e4bf95\n(cherry picked from commit 8fa3de4a508535085fcabc8872b38894619dfa98)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/d69b8e9afc14b0c83cf066dd3b1ccd26526460c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/d69b8e9afc14b0c83cf066dd3b1ccd26526460c1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
