)]}'
{"id":"openstack%2Fsahara-plugin-cdh~657923","triplet_id":"openstack%2Fsahara-plugin-cdh~master~I5047c85fbbf554542ca5655d7c9315484f9cfbb4","project":"openstack/sahara-plugin-cdh","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I5047c85fbbf554542ca5655d7c9315484f9cfbb4","subject":"Update Python 3 test runtimes for Train","status":"MERGED","created":"2019-05-09 06:26:37.000000000","updated":"2019-05-16 17:17:26.000000000","submitted":"2019-05-16 17:17:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657923-1558027046949-4a183efd","meta_rev_id":"a92cffbda0df9ce9583045823c86a81711fe3fd2","_number":657923,"virtual_id_number":657923,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2019-05-16 17:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2019-05-09 12:33:08.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":1,"date":"2019-05-10 11:16:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-05-16 16:56:19.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-05-16 16:56:19.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:33:08.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-05-10 11:16:55.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-05-16 16:56:19.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-05-16 17:17:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7bd983be9461ef6301ba00c2bffa4ef91501d85","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-09 06:26:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbaf1bc18c851593001d2e491044e876e89f5f12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 07:07:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/23/657923/1/check/test-release-openstack/cacd5ce/ : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/657923/1/check/openstack-tox-lower-constraints/e0dfe77/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/23/657923/1/check/openstack-tox-pep8/4e28a29/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/23/657923/1/check/openstack-tox-py27/5723910/ : SUCCESS in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/23/657923/1/check/openstack-tox-py36/2e15b01/ : SUCCESS in 6m 08s\n- openstack-tox-py37 http://logs.openstack.org/23/657923/1/check/openstack-tox-py37/b2e56df/ : SUCCESS in 7m 07s\n- openstack-tox-docs http://logs.openstack.org/23/657923/1/check/openstack-tox-docs/0852ec3/html/ : SUCCESS in 6m 04s\n- build-openstack-releasenotes http://logs.openstack.org/23/657923/1/check/build-openstack-releasenotes/b791f50/html/ : SUCCESS in 5m 01s\n- sahara-buildimages-cloudera http://logs.openstack.org/23/657923/1/check/sahara-buildimages-cloudera/c91930e/ : FAILURE in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5df13a1b0ad5f5c940c21d6c087d81feda09ac0a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-09 09:36:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2f85653ecd4d0688f71fc64938fa67d1d4a699aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 09:46:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/23/657923/2/check/test-release-openstack/0774d92/ : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/657923/2/check/openstack-tox-lower-constraints/6327a46/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/23/657923/2/check/openstack-tox-pep8/75d8cff/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/23/657923/2/check/openstack-tox-py27/2cb8f8a/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/23/657923/2/check/openstack-tox-py36/60306f4/ : SUCCESS in 5m 06s\n- openstack-tox-py37 http://logs.openstack.org/23/657923/2/check/openstack-tox-py37/eeaff2a/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/23/657923/2/check/openstack-tox-docs/3566214/html/ : SUCCESS in 4m 58s\n- build-openstack-releasenotes http://logs.openstack.org/23/657923/2/check/build-openstack-releasenotes/3656355/html/ : SUCCESS in 3m 41s\n- sahara-buildimages-cloudera http://logs.openstack.org/23/657923/2/check/sahara-buildimages-cloudera/889a73a/ : FAILURE in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1b7f32d71745cdf1f55cf42b89ec5304cfa7af9","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-05-09 12:33:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"13a8dbbaeb993449e57b0202016a544473e79766","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-10 11:16:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"904372baf5ba9a1d0c2661c2bf7d73f0c6fb4163","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-05-16 16:56:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ac1f4775ff0bae7271279b5b5bf43d762d534c28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:56:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7dd6e0859a997855a6bddf9497ee0f8e0acd2c5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 17:17:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f3abdd6ec121b0f5839713fff4be683c0f0c22b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 17:17:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/23/657923/2/gate/test-release-openstack/be8703f/ : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/657923/2/gate/openstack-tox-lower-constraints/3f5bb5c/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/23/657923/2/gate/openstack-tox-pep8/1aa9d64/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/23/657923/2/gate/openstack-tox-py27/7039437/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/23/657923/2/gate/openstack-tox-py36/7fa7c20/ : SUCCESS in 4m 49s\n- openstack-tox-py37 http://logs.openstack.org/23/657923/2/gate/openstack-tox-py37/622c05a/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/23/657923/2/gate/openstack-tox-docs/cd624d8/html/ : SUCCESS in 5m 02s\n- build-openstack-releasenotes http://logs.openstack.org/23/657923/2/gate/build-openstack-releasenotes/fd2a5c9/html/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e44a62db75d26b97a4917b1c288b7a0b6dc660a","revisions":{"7d77b9e5b087d7c57176391e3f3d9c2f9df09ca1":{"kind":"REWORK","_number":1,"created":"2019-05-09 06:26:37.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/23/657923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/23/657923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/23/657923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/23/657923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/23/657923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/23/657923/1"}}},"commit":{"parents":[{"commit":"8fa3de4a508535085fcabc8872b38894619dfa98","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/8fa3de4a508535085fcabc8872b38894619dfa98"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:00.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:00.000000000","tz":480},"subject":"Update Python 3 test runtimes for Train","message":"Update Python 3 test runtimes for Train\n\nFollowing the process set down in the TC resolution on updating Python\nversions:\n\nhttps://governance.openstack.org/tc/resolutions/20181024-python-update-process.html\n\nChange-Id: I5047c85fbbf554542ca5655d7c9315484f9cfbb4\nDepends-On: https://review.opendev.org/#/c/641878/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/7d77b9e5b087d7c57176391e3f3d9c2f9df09ca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/7d77b9e5b087d7c57176391e3f3d9c2f9df09ca1"}]},"branch":"refs/heads/master"},"6e44a62db75d26b97a4917b1c288b7a0b6dc660a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-09 09:36:01.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/23/657923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/23/657923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/23/657923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/23/657923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/23/657923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/23/657923/2"}}},"commit":{"parents":[{"commit":"8fa3de4a508535085fcabc8872b38894619dfa98","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/8fa3de4a508535085fcabc8872b38894619dfa98"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:00.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 09:34:36.000000000","tz":480},"subject":"Update Python 3 test runtimes for Train","message":"Update Python 3 test runtimes for Train\n\nThis goal is to implement the process set out in the 2018-10-24 Python\nUpdate Process TC resolution[1], for the Train cycle to ensure unit\ntesting is in place for all of the Tested Runtimes for Train[2].\nIn practice, this generally means adding unit tests for Python 3.7 and dropping\nunit tests for Python 3.5. Using the Zuul template for Train will ensure that\nall projects that support Python3 will be tested against the agreed runtime\nversions, and make it easier to update them in future.\n\n[1]https://governance.openstack.org/tc/resolutions/20181024-python-update-process.html\n[2]https://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I5047c85fbbf554542ca5655d7c9315484f9cfbb4\nDepends-On: https://review.opendev.org/#/c/641878/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/6e44a62db75d26b97a4917b1c288b7a0b6dc660a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/6e44a62db75d26b97a4917b1c288b7a0b6dc660a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
