)]}'
{"id":"openstack%2Fsahara-plugin-cdh~667306","triplet_id":"openstack%2Fsahara-plugin-cdh~master~I7a378c69f3ac35ec398400ff8be8421b619a8e6b","project":"openstack/sahara-plugin-cdh","branch":"master","topic":"sphinx-reqs","hashtags":[],"change_id":"I7a378c69f3ac35ec398400ff8be8421b619a8e6b","subject":"Update sphinx from current requirements","status":"MERGED","created":"2019-06-25 09:47:09.000000000","updated":"2019-06-25 17:37:50.000000000","submitted":"2019-06-25 17:37:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667306-1561484270888-f6af9c4d","meta_rev_id":"7b720b8dcc73b5bb02b6e8560929912b43f579df","_number":667306,"virtual_id_number":667306,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":2,"date":"2019-06-25 17:37:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2019-06-25 10:44:54.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2019-06-25 17:27:57.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":1,"date":"2019-06-25 17:27:57.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 10:44:54.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-06-25 17:27:57.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-06-25 17:37:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a003953552b3ac0f3a3e4070d38df41e14c2c5bc","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-06-25 09:47:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f837023df72194d1c5783259fea30240e6b1093","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 10:21:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/06/667306/1/check/requirements-check/f8c1946/ : SUCCESS in 2m 10s\n- openstack-tox-pep8 http://logs.openstack.org/06/667306/1/check/openstack-tox-pep8/a44b656/ : SUCCESS in 3m 38s\n- openstack-tox-docs http://logs.openstack.org/06/667306/1/check/openstack-tox-docs/0ec1a58/html/ : SUCCESS in 3m 27s\n- build-openstack-releasenotes http://logs.openstack.org/06/667306/1/check/build-openstack-releasenotes/66c5c0f/html/ : SUCCESS in 4m 17s\n- sahara-buildimages-cloudera http://logs.openstack.org/06/667306/1/check/sahara-buildimages-cloudera/4b098f1/ : FAILURE in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6666988e8f481fadfa9c46fa84bd0198a484fcff","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-06-25 10:44:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9a51bf44add1fe0fa9fd17fc47ed281f3c2492b0","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-06-25 17:27:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"961c32431971bcdcad8fbaff872d66a564269b62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 17:28:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"968fd805394f3d85b5406ab920a28224108b6bc3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 17:37:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"477982b44c99354e98b7df9b3997676d92f20d61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 17:37:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/06/667306/1/gate/requirements-check/058afc2/ : SUCCESS in 2m 20s\n- openstack-tox-pep8 http://logs.openstack.org/06/667306/1/gate/openstack-tox-pep8/d328c0c/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/06/667306/1/gate/openstack-tox-docs/4b70e5b/html/ : SUCCESS in 3m 17s\n- build-openstack-releasenotes http://logs.openstack.org/06/667306/1/gate/build-openstack-releasenotes/f8986a0/html/ : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f956bbe90e1db1ba3f6f35ad485225f2d12d141","revisions":{"4f956bbe90e1db1ba3f6f35ad485225f2d12d141":{"kind":"REWORK","_number":1,"created":"2019-06-25 09:47:09.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/06/667306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/06/667306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/06/667306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/06/667306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/06/667306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/06/667306/1"}}},"commit":{"parents":[{"commit":"6e44a62db75d26b97a4917b1c288b7a0b6dc660a","subject":"Update Python 3 test runtimes for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/6e44a62db75d26b97a4917b1c288b7a0b6dc660a"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-06-25 09:43:09.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-06-25 09:46:35.000000000","tz":120},"subject":"Update sphinx from current requirements","message":"Update sphinx from current requirements\n\nOtherwise the requirements-check job fails.\n\nChange-Id: I7a378c69f3ac35ec398400ff8be8421b619a8e6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/4f956bbe90e1db1ba3f6f35ad485225f2d12d141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/4f956bbe90e1db1ba3f6f35ad485225f2d12d141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
