)]}'
{"id":"openstack%2Fsahara-plugin-cdh~673892","triplet_id":"openstack%2Fsahara-plugin-cdh~master~I5de5a2109f0b50adc209008738920fb11fd0fd76","project":"openstack/sahara-plugin-cdh","branch":"master","hashtags":[],"change_id":"I5de5a2109f0b50adc209008738920fb11fd0fd76","subject":"Fixes on CDH for python 3 compatibility","status":"MERGED","created":"2019-07-31 18:25:45.000000000","updated":"2019-10-04 14:16:56.000000000","submitted":"2019-10-04 14:15:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673892-1570198549056-6aa585c6","meta_rev_id":"6b79a48daba62fc7defd6dfdc26dceacf6cd24d3","_number":673892,"virtual_id_number":673892,"owner":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-03 19:01:38.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 14:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2019-10-04 14:01:44.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":2,"date":"2019-10-04 12:58:59.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":1,"date":"2019-10-04 14:01:44.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 12:58:59.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-10-04 14:01:44.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-10-04 14:15:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a73e6d80d3e75c33cf1ce11165ea0f16d8fab7e","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-07-31 18:25:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6cd4954b601faac8aea52e4548715639bf96b6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 20:32:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/92/673892/1/check/openstack-tox-lower-constraints/ef9cb0a/ : FAILURE in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/92/673892/1/check/openstack-tox-pep8/0d89bf7/ : FAILURE in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/92/673892/1/check/openstack-tox-py27/6b6b26a/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/92/673892/1/check/openstack-tox-py36/095fddc/ : FAILURE in 6m 20s\n- openstack-tox-py37 http://logs.openstack.org/92/673892/1/check/openstack-tox-py37/6af9d8f/ : FAILURE in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/92/673892/1/check/openstack-tox-docs/983f159/html/ : SUCCESS in 4m 20s\n- sahara-buildimages-cloudera http://logs.openstack.org/92/673892/1/check/sahara-buildimages-cloudera/256c5e2/ : TIMED_OUT in 2h 00m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9682e89a99656e385b7d1b347bfdfdf197b03b16","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-02 12:48:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f29509d4a7ab6fa71744f962794b58cc9bae454d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 14:54:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/92/673892/2/check/openstack-tox-lower-constraints/e6fa278/ : FAILURE in 4m 42s\n- openstack-tox-pep8 https://logs.opendev.org/92/673892/2/check/openstack-tox-pep8/a7ff966/ : FAILURE in 5m 10s\n- openstack-tox-py27 https://logs.opendev.org/92/673892/2/check/openstack-tox-py27/6b2c1e7/ : SUCCESS in 5m 11s\n- openstack-tox-py36 https://logs.opendev.org/92/673892/2/check/openstack-tox-py36/d932517/ : FAILURE in 4m 53s\n- openstack-tox-py37 https://logs.opendev.org/92/673892/2/check/openstack-tox-py37/fa09c56/ : FAILURE in 5m 12s\n- openstack-tox-docs https://logs.opendev.org/92/673892/2/check/openstack-tox-docs/8835f12/html/ : SUCCESS in 4m 34s\n- sahara-buildimages-cloudera https://logs.opendev.org/92/673892/2/check/sahara-buildimages-cloudera/ab39536/ : TIMED_OUT in 2h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3dab95a0e792e6267cd247a0570f6c0eea6167e","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-07 13:31:00.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"242897d042cb4f416137c5e031edee958a87924c","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-07 13:31:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"4367755d9a7e4643f54c1c6616dcdbbc85a32774","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-07 13:35:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4cf7032994a02d6fa3e8cdef1dfc5fba3e2a2f59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 15:38:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/92/673892/5/check/openstack-tox-lower-constraints/360d14a/ : FAILURE in 6m 18s\n- openstack-tox-pep8 https://logs.opendev.org/92/673892/5/check/openstack-tox-pep8/db5e4bb/ : SUCCESS in 4m 32s\n- openstack-tox-py27 https://logs.opendev.org/92/673892/5/check/openstack-tox-py27/a96d549/ : SUCCESS in 3m 44s\n- openstack-tox-py36 https://logs.opendev.org/92/673892/5/check/openstack-tox-py36/9234c08/ : FAILURE in 5m 39s\n- openstack-tox-py37 https://logs.opendev.org/92/673892/5/check/openstack-tox-py37/33fd43d/ : FAILURE in 5m 08s\n- openstack-tox-docs https://logs.opendev.org/92/673892/5/check/openstack-tox-docs/4f6fdaa/html/ : SUCCESS in 3m 44s\n- sahara-buildimages-cloudera https://logs.opendev.org/92/673892/5/check/sahara-buildimages-cloudera/f938a39/ : TIMED_OUT in 2h 00m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d129b126e31fec044699b4001bbaf3fa99225ad5","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-08 13:00:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"73fad6d57de9354eac9b764597c1c225e47716a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:08:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/92/673892/6/check/openstack-tox-lower-constraints/eb91e63/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://logs.opendev.org/92/673892/6/check/openstack-tox-pep8/aa343e4/ : SUCCESS in 4m 58s\n- openstack-tox-py27 https://logs.opendev.org/92/673892/6/check/openstack-tox-py27/38c38a2/ : SUCCESS in 4m 50s\n- openstack-tox-py36 https://logs.opendev.org/92/673892/6/check/openstack-tox-py36/274c7e6/ : SUCCESS in 3m 44s\n- openstack-tox-py37 https://logs.opendev.org/92/673892/6/check/openstack-tox-py37/227a768/ : SUCCESS in 4m 25s\n- openstack-tox-docs https://logs.opendev.org/92/673892/6/check/openstack-tox-docs/11077ea/html/ : SUCCESS in 4m 26s\n- sahara-buildimages-cloudera https://logs.opendev.org/92/673892/6/check/sahara-buildimages-cloudera/efecf2d/ : TIMED_OUT in 2h 00m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"860c6bb137188328a1897c69c0a83492c8944c37","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-14 11:52:47.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"951c9cb99704f657be6bbe58a8cd454db5f786f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 13:58:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f9033cd502845c5a997ceeebf37bf4e : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab658c0cb4e4ec697223468bbcc1f52 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1898635a3bfa47c4bcc5eac4ac48acb0 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d39225f9b6dd431682432c4ff2e5930d : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09b81772f55c4c85bf786a2a618be24b : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9c937b0c4f54d659aa6f78062ec5b93 : SUCCESS in 2m 49s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/8a033b5fb3ff4d0a8a18c105eeab71cf : TIMED_OUT in 2h 00m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9cf0cc03f50fdbd3c381ee4dd46890089a3aae11","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-20 18:08:47.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"8f6fbc776ff3afcb832e95b15882f87de550d2e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 20:16:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23e4b709f2484663b03aaf9b5692d6ec : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/000c9bb9070f46e4842fd64d69023f3e : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/504581b0ecc945419d325320c31cc610 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39228bac81f54053840bf9927da37ea0 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6197debae594ebc841a25ad781c35cb : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d635de8a3154bb0ba7d67733f032ff5 : SUCCESS in 5m 31s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/50244070cde14d4b925317e4ae8b7a14 : TIMED_OUT in 2h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"08d37fc3b95a4548f52d3d055ba8a75388d5e6f3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-08-28 14:03:52.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI have an old comment that I failed to send, but it may be still relevant:","accounts_in_message":[],"_revision_number":2},{"id":"64f620344e358d905f0798559cfa3b1bdfecbc26","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-28 16:06:25.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"f8b5794a706b30f9cabdaaa58b80c4d85f19e5f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 20:22:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4a3830f293a4d669a2b1b2f9f666ce2 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791651e7119d40bf86834aae432853ff : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96e741528e76443e9459373d2bacf3a0 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d5c6b76e1f340b79656657609d79b3f : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58a3b71d314c4af29868e12ac916f271 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c18a59226a404297bfc31e9ef69cc4 : SUCCESS in 5m 06s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/79ec39914bd941c1804cfac16c4fd853 : TIMED_OUT in 2h 01m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"73ecb5a49c9d992ead94c3e4a7ca3f89dfcde98e","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-08-30 14:22:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c9b49f961465b0c4f316b5f0df23d8799532502","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-30 14:40:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff215138ca27e634606f54a03ae1f2fb8617641b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-03 09:31:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf5d0efbc845bb78a8c0dad6db8f43628b9ff50a","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-10-03 19:01:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"caacf973e0f1633310955860923a57772e0482cc","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-10-03 19:03:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6fec0c57f1a8ed894fba863725cdd2d65de92f9c","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-10-03 19:04:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18df5ba109b90579d438d76bdba167ae70fca811","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-04 12:58:59.000000000","message":"Patch Set 8: Code-Review+2\n\nOk, so let\u0027s go.","accounts_in_message":[],"_revision_number":8},{"id":"95a15b9bd6c8da075442745326fbc7133bc1a5a0","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-10-04 14:01:44.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1746d10bce5b4027b8cb23ed9cd72436959f37aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:02:00.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7e4861c14802368468da3f581b9175accc86e053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:15:48.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4872785af524673b3dfceb8c06ba058 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68739615f44649819e469113fb9ea8f4 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47002b90400240a6876c84ae2bc482f2 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/222d32bf2e12401197d5dbf9b17cd3e5 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c474f430f3f441ebcde876b6d553494 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bea9ecadd6b403587d3bc2752663e67 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":8},{"id":"179d8f4b5531d07cde656185275125621e1488cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:15:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"6b79a48daba62fc7defd6dfdc26dceacf6cd24d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:16:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8071230fbde846ddb9113e4b23945a72 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"9d443815fcfdd3fe11b6cbe3c47660920a2a8fc5","revisions":{"be76adc13ed5a7ab8146aae8280cbbff97e25c0e":{"kind":"REWORK","_number":1,"created":"2019-07-31 18:25:45.000000000","uploader":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"ref":"refs/changes/92/673892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/92/673892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/1"}}},"commit":{"parents":[{"commit":"21a54be52f61307d508f3a9395c7a67a76388228","subject":"Merge \"Migrate away from oslo_i18n.enable_lazy()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/21a54be52f61307d508f3a9395c7a67a76388228"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"committer":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"subject":"Fixes on CDH for python 3 compatibility","message":"Fixes on CDH for python 3 compatibility\n\nChange-Id: I5de5a2109f0b50adc209008738920fb11fd0fd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/be76adc13ed5a7ab8146aae8280cbbff97e25c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/be76adc13ed5a7ab8146aae8280cbbff97e25c0e"}]},"branch":"refs/heads/master"},"1d522cc173cc9cee9d628285ce677a4b5e35564b":{"kind":"REWORK","_number":2,"created":"2019-08-02 12:48:01.000000000","uploader":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"ref":"refs/changes/92/673892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/92/673892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/2"}}},"commit":{"parents":[{"commit":"21a54be52f61307d508f3a9395c7a67a76388228","subject":"Merge \"Migrate away from oslo_i18n.enable_lazy()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/21a54be52f61307d508f3a9395c7a67a76388228"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"committer":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-08-02 12:47:42.000000000","tz":-180},"subject":"Fixes on CDH for python 3 compatibility","message":"Fixes on CDH for python 3 compatibility\n\nStory: #2005914\nTask: #34174\nChange-Id: I5de5a2109f0b50adc209008738920fb11fd0fd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/1d522cc173cc9cee9d628285ce677a4b5e35564b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/1d522cc173cc9cee9d628285ce677a4b5e35564b"}]},"branch":"refs/heads/master"},"cb837790a538fc6cb41981c24946f452f40c442f":{"kind":"NO_CHANGE","_number":3,"created":"2019-08-07 13:31:00.000000000","uploader":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"ref":"refs/changes/92/673892/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/92/673892/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/3"}}},"commit":{"parents":[{"commit":"21a54be52f61307d508f3a9395c7a67a76388228","subject":"Merge \"Migrate away from oslo_i18n.enable_lazy()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/21a54be52f61307d508f3a9395c7a67a76388228"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"committer":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-08-07 13:30:35.000000000","tz":-180},"subject":"Fixes on CDH for python 3 compatibility","message":"Fixes on CDH for python 3 compatibility\n\nStory: #2005914\nTask: #34174\nChange-Id: I5de5a2109f0b50adc209008738920fb11fd0fd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/cb837790a538fc6cb41981c24946f452f40c442f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/cb837790a538fc6cb41981c24946f452f40c442f"}]},"branch":"refs/heads/master"},"5bf241f80faf1b1975aa712fabe4b7b702ae4ab6":{"kind":"NO_CHANGE","_number":4,"created":"2019-08-07 13:31:50.000000000","uploader":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"ref":"refs/changes/92/673892/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/92/673892/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/4"}}},"commit":{"parents":[{"commit":"21a54be52f61307d508f3a9395c7a67a76388228","subject":"Merge \"Migrate away from oslo_i18n.enable_lazy()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/21a54be52f61307d508f3a9395c7a67a76388228"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"committer":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-08-07 13:31:41.000000000","tz":-180},"subject":"Fixes on CDH for python 3 compatibility","message":"Fixes on CDH for python 3 compatibility\n\nStory: #2005914\nTask: #34174\nChange-Id: I5de5a2109f0b50adc209008738920fb11fd0fd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/5bf241f80faf1b1975aa712fabe4b7b702ae4ab6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/5bf241f80faf1b1975aa712fabe4b7b702ae4ab6"}]},"branch":"refs/heads/master"},"271aca1592033cf37504e6861bb8ca486f45a507":{"kind":"REWORK","_number":5,"created":"2019-08-07 13:35:53.000000000","uploader":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"ref":"refs/changes/92/673892/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/92/673892/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/5"}}},"commit":{"parents":[{"commit":"21a54be52f61307d508f3a9395c7a67a76388228","subject":"Merge \"Migrate away from oslo_i18n.enable_lazy()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/21a54be52f61307d508f3a9395c7a67a76388228"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"committer":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-08-07 13:35:28.000000000","tz":-180},"subject":"Fixes on CDH for python 3 compatibility","message":"Fixes on CDH for python 3 compatibility\n\nStory: #2005914\nTask: #34174\nChange-Id: I5de5a2109f0b50adc209008738920fb11fd0fd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/271aca1592033cf37504e6861bb8ca486f45a507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/271aca1592033cf37504e6861bb8ca486f45a507"}]},"branch":"refs/heads/master"},"1f09daf9c4ad9d93ea0b4bd80f401a2118f81df5":{"kind":"REWORK","_number":6,"created":"2019-08-08 13:00:34.000000000","uploader":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"ref":"refs/changes/92/673892/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/92/673892/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/6"}}},"commit":{"parents":[{"commit":"21a54be52f61307d508f3a9395c7a67a76388228","subject":"Merge \"Migrate away from oslo_i18n.enable_lazy()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/21a54be52f61307d508f3a9395c7a67a76388228"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"committer":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-08-08 13:00:23.000000000","tz":-180},"subject":"Fixes on CDH for python 3 compatibility","message":"Fixes on CDH for python 3 compatibility\n\nStory: #2005914\nTask: #34174\nChange-Id: I5de5a2109f0b50adc209008738920fb11fd0fd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/1f09daf9c4ad9d93ea0b4bd80f401a2118f81df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/1f09daf9c4ad9d93ea0b4bd80f401a2118f81df5"}]},"branch":"refs/heads/master"},"8f4be8759f52a81ead33fe80cca05b1d8d14a306":{"kind":"NO_CHANGE","_number":7,"created":"2019-08-14 11:52:47.000000000","uploader":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"ref":"refs/changes/92/673892/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/92/673892/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/7"}}},"commit":{"parents":[{"commit":"21a54be52f61307d508f3a9395c7a67a76388228","subject":"Merge \"Migrate away from oslo_i18n.enable_lazy()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/21a54be52f61307d508f3a9395c7a67a76388228"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"committer":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-08-14 11:52:37.000000000","tz":-180},"subject":"Fixes on CDH for python 3 compatibility","message":"Fixes on CDH for python 3 compatibility\n\nStory: #2005914\nTask: #34174\nChange-Id: I5de5a2109f0b50adc209008738920fb11fd0fd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/8f4be8759f52a81ead33fe80cca05b1d8d14a306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/8f4be8759f52a81ead33fe80cca05b1d8d14a306"}]},"branch":"refs/heads/master"},"9d443815fcfdd3fe11b6cbe3c47660920a2a8fc5":{"kind":"NO_CHANGE","_number":8,"created":"2019-08-20 18:08:47.000000000","uploader":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"ref":"refs/changes/92/673892/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/92/673892/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/92/673892/8"}}},"commit":{"parents":[{"commit":"21a54be52f61307d508f3a9395c7a67a76388228","subject":"Merge \"Migrate away from oslo_i18n.enable_lazy()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/21a54be52f61307d508f3a9395c7a67a76388228"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"committer":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-08-20 18:08:39.000000000","tz":-180},"subject":"Fixes on CDH for python 3 compatibility","message":"Fixes on CDH for python 3 compatibility\n\nStory: #2005914\nTask: #34174\nChange-Id: I5de5a2109f0b50adc209008738920fb11fd0fd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/9d443815fcfdd3fe11b6cbe3c47660920a2a8fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/9d443815fcfdd3fe11b6cbe3c47660920a2a8fc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
