)]}'
{"id":"openstack%2Fsahara-plugin-cdh~679249","triplet_id":"openstack%2Fsahara-plugin-cdh~master~I353e33db27c8d676398973995960b98d3163aa0d","project":"openstack/sahara-plugin-cdh","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"I353e33db27c8d676398973995960b98d3163aa0d","subject":"PDF documentation build","status":"MERGED","created":"2019-08-29 13:32:24.000000000","updated":"2019-09-20 17:19:31.000000000","submitted":"2019-09-20 17:17:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679249-1568999877727-b80b591b","meta_rev_id":"f3d49001cd6b90fd28847685eb646334e5e3b24a","_number":679249,"virtual_id_number":679249,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2019-09-20 17:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-20 17:02:48.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2019-09-19 15:36:55.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":1,"date":"2019-09-14 15:44:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-13 10:36:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-20 17:02:48.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 04:57:33.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-13 10:36:22.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-09-14 15:44:17.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-09-19 15:36:55.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-09-20 17:02:48.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-09-20 17:17:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd658b748e8e9a9d4f4b68626de369843a219fd9","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-08-29 13:32:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"536d294678830b68e81abc53e3c3f7462a321500","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-29 13:34:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dca7af74630c46a7845feaca3a10ad6c29fa68cf","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-08-29 13:40:08.000000000","message":"Patch Set 1:\n\nUhm, at least the \u0027 signs in \"can\u0027t\" are not rendered in my local testing, let\u0027s see here.","accounts_in_message":[],"_revision_number":1},{"id":"b93aa24366ba2a0a7f2b2ecf7cba7bfa00adee36","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-08-29 13:53:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e1071cc63c93f3df3a1b3c56ad02ece9946a018d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 14:46:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2615bc00c85741f68188efee418ac778 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723bd3a30aaa491b97d3487d0944f1b9 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5baaf364bf684faf9b9b5d24369a4084 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4687ff419f414ea7a36bd7b2a864d721 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/911561994b91421e883697bc6b27a672 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97681a98ea854fdba66e25ce6fdfa5d6 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a305fac69d75472daeebd703a4d67b7b : SUCCESS in 3m 44s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/161e1f127c9d49328281115f83a6486c : FAILURE in 44m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a19afe2810ded154f348e172756174cd5d02d3c6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 04:57:33.000000000","message":"Patch Set 2: Code-Review-1\n\nNote that https://review.opendev.org/679777 changed the pdf file name to doc-PROJECT.pdf, please adjust doc/source/conf.py","accounts_in_message":[],"_revision_number":2},{"id":"215eebc8e9aacc6b5e964e9f857588c88bf02c5a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-09-06 14:15:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eebebce46bc3147069c7e4c9cb7ea2af21e1fb63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 16:31:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4b467e1d27b4ab9af78a76264f96e5f : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfdccd3ea04a4f9682e56084e77b0f0f : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ef5cfc5070a48f8a03b02f77601cfa7 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e9f44d3f4ac45beac549b6525f581a6 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88d98a35eade41978aa198b08ea51d9e : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372c7050296d4b219efe6c1d92b56e00 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2796389ff2845a5ae16d5adacafee58 : SUCCESS in 3m 20s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/5ac4ec23499e4eefb87b34e49c653741 : TIMED_OUT in 2h 00m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f00cc2d4511d5708debf54959e18a0fdf564b69","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-13 10:36:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0a8d3072ae4e8a5116aa2dfd91248cb992b5065d","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-09-14 15:44:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c85db1dcb42673fd938bb541fd079cbd627ea557","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-09-19 15:36:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fe1c7469b4ec8a33ddbf25ed7305d405c55ed42c","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-09-20 17:02:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a74b1aae05b2d597e06137f33045f37dfc30fbc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 17:03:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a64d92e2cb7c98e32d70b8f7ead2de52e2f26e52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 17:17:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25b3fa03eaf7434a932e517430d2ffe8 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/355f36b22105489cb40e63800fbe080c : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd2d17baae7a4e57b81d6dd6a7253f61 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2922c3a0ad34cb497037b346289a65f : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fbf0dd7e47c4ee480f9988eda0f058e : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e7bce23b9a471e8a67a83b7eb6d5ab : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7505b7fbb60e40faa74d8710bc952c98 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":3},{"id":"d019fe74aa19dbef5e75ccd7eee912bdda8be100","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 17:17:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"117611f23f4ba50fe9a5c8deacfce0003d5fbb40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 17:19:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2484a299f6dd45acb581e42d8251b3c7 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/406d9010ae9f45369b021edd1abf47b3 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bc4673f3b9d3ec82f87b8ec0ba58c9ed5f4025b9","revisions":{"171acb689187a2fdd15662f00d7d0d2cb139fca3":{"kind":"REWORK","_number":1,"created":"2019-08-29 13:32:24.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/49/679249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/49/679249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/1"}}},"commit":{"parents":[{"commit":"56e15f929e95714cea8674a547bf5f38fba8728b","subject":"Limit envlist to py37 for Python 3 Train goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/56e15f929e95714cea8674a547bf5f38fba8728b"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-08-29 13:23:13.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-08-29 13:26:35.000000000","tz":120},"subject":"PDF documentation build","message":"PDF documentation build\n\n- add the pdf-docs tox environment according the established guidelines;\n- remove a useless extra title (for both builders);\n- docs tox environment: clean up only the generated html, not the entire\n  build directory;\n- improve the title of the latex document.\n\nChange-Id: I353e33db27c8d676398973995960b98d3163aa0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/171acb689187a2fdd15662f00d7d0d2cb139fca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/171acb689187a2fdd15662f00d7d0d2cb139fca3"}]},"branch":"refs/heads/master"},"27fa3ca3f23876399f1273d21edb7dc04efb8ea6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-29 13:53:53.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/49/679249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/49/679249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/2"}}},"commit":{"parents":[{"commit":"56e15f929e95714cea8674a547bf5f38fba8728b","subject":"Limit envlist to py37 for Python 3 Train goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/56e15f929e95714cea8674a547bf5f38fba8728b"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-08-29 13:53:40.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-08-29 13:53:40.000000000","tz":120},"subject":"PDF documentation build","message":"PDF documentation build\n\n- add the pdf-docs tox environment according the established guidelines;\n- remove a useless extra title (for both builders);\n- docs tox environment: clean up only the generated html, not the entire\n  build directory;\n- improve the title of the latex document.\n\nStory: 2006117\nTask: 35491\nChange-Id: I353e33db27c8d676398973995960b98d3163aa0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/27fa3ca3f23876399f1273d21edb7dc04efb8ea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/27fa3ca3f23876399f1273d21edb7dc04efb8ea6"}]},"branch":"refs/heads/master"},"bc4673f3b9d3ec82f87b8ec0ba58c9ed5f4025b9":{"kind":"REWORK","_number":3,"created":"2019-09-06 14:15:00.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/49/679249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/49/679249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/679249/3"}}},"commit":{"parents":[{"commit":"3634375600a58d4aab4e67c9f02fba256d4fd650","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/3634375600a58d4aab4e67c9f02fba256d4fd650"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-09-06 14:09:53.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-09-06 14:14:15.000000000","tz":120},"subject":"PDF documentation build","message":"PDF documentation build\n\n- add the pdf-docs tox environment according the established guidelines;\n- remove a useless extra title (for both builders);\n- docs tox environment: clean up only the generated html, not the entire\n  build directory;\n- disable the too smart smartquotes option for the latex builder;\n- improve the title of the latex document.\n\nStory: 2006117\nTask: 35491\nChange-Id: I353e33db27c8d676398973995960b98d3163aa0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/bc4673f3b9d3ec82f87b8ec0ba58c9ed5f4025b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/bc4673f3b9d3ec82f87b8ec0ba58c9ed5f4025b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
